)]}'
{"id":"openstack%2Fpuppet-glance~894252","triplet_id":"openstack%2Fpuppet-glance~master~Ibb12bdb8333fc91e2dc95c2c2cc4de79f2c82cc1","project":"openstack/puppet-glance","branch":"master","topic":"bug/2034893","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-08 14:25:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibb12bdb8333fc91e2dc95c2c2cc4de79f2c82cc1","subject":"Fix legacy backend class usage with multi_store\u003dfalse","status":"MERGED","created":"2023-09-08 05:27:55.000000000","updated":"2023-09-08 14:25:25.000000000","submitted":"2023-09-08 14:25:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894252-bug/2034893","meta_rev_id":"1d2c94027296e343f5b0d889e24eda9e69f43544","_number":894252,"virtual_id_number":894252,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 14:25:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-09-08 08:26:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-09-08 13:00:16.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-09-08 13:00:16.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-08 07:47:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-08 08:26:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-09-08 13:00:16.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"9ec3721d6757256817f6af93b17bdaec2d455b23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 05:27:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3af1c76c805ad5d8a6b4ab9e9acb9460cec676a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 05:33:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a41f67cd08a92ba5e356ac7752253dcadbd95883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 06:30:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7d2a8e0cac513b9cd99846eea368084498b2bf3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 07:47:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0edcd301b89049548e1d511f2573e89e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6b10fc63d6149c2b6ad9c6e38180a6c : SUCCESS in 3m 13s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de8a40c83255434c9fd0dae339ea2a18 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a37752d314c4b96baa5a4d6f0abe80d : SUCCESS in 5m 30s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/30e186b1292144f59b5595877fb803b2 : SUCCESS in 11m 49s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83de979d516f427cbbde1f10a8dd98a6 : SUCCESS in 11m 02s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0dade071bbf41ab94cd53f102879b87 : SUCCESS in 11m 50s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c986640cce494a1bb2fc9a819ebc7467 : SUCCESS in 8m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c0f03e7f62c4d619f490fc0f8eb7a99 : FAILURE in 44m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1ae07aee4db4a0abdb87ab02c99d61b : SUCCESS in 47m 48s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53ff8daaf846422593233347af164f8e : SUCCESS in 39m 56s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97abacf3f4624d2284bff70127ac55d5 : SUCCESS in 46m 53s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95a533d9ac60412b8a33943e04714b39 : SUCCESS in 40m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2d989dcf50748d093cad6084e77a209 : SUCCESS in 1h 07m 16s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/634ba5cfc42b4f699718ac1985455a8f : SUCCESS in 1h 11m 25s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1250f6193f2244deb7cf5a6f88a3abbb : SUCCESS in 1h 04m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f7f18a8764c44dab6491a541888ccf6 : SUCCESS in 47m 33s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4298ee2f10124b87848d1e3c11493656 : SUCCESS in 58m 26s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d0a39429667467abf4a5ef1e3ff14a9 : SUCCESS in 16m 01s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/112e63ac4ba44a1f98855fc83f7f524f : SUCCESS in 13m 58s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/302ff8c2bc6a48f4ac25852fbfb8b7a9 : SUCCESS in 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"792985e0fd0bd95b762fb0d52bb45c21f3181bfa","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-09-08 08:26:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4172b69da8bf356d76f957a4f6f8b0d70ac154ce","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-09-08 13:00:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"08ddcefd293049545a9f5e9c50c0532de67235f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:01:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"34b27e1352e9f7622a544654801fbeb70eeffc5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 14:25:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7641099edb74e26be8f02a088286aa8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7cd88d50a05499d82dc5ca99e049cce : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f3af6c340094298a00b94c941e9f213 : SUCCESS in 2m 54s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9596e260eb0648a295de43d9aba9453f : SUCCESS in 13m 04s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/949978375ca044c5b967085e5f10965c : SUCCESS in 8m 49s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/d817ab67ed0244e696f71426dc999cfc : SUCCESS in 11m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/05c21ab1707d440699546e63ffec5d71 : SUCCESS in 1h 03m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a0c0dbdbcd0449bbe5de7fe57c44d6e : SUCCESS in 1h 19m 43s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/85945b93bef640d9b1be99a790d6ec26 : SUCCESS in 1h 06m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/27f14c709c804624af18e69ab57963ae : SUCCESS in 59m 54s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1b3e821be1a45738c4f2fa5a0e20b0b : SUCCESS in 56m 17s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c5ea82ae6364d9f9bcf63da3f176c36 : SUCCESS in 20m 17s","accounts_in_message":[],"_revision_number":3},{"id":"1d2c94027296e343f5b0d889e24eda9e69f43544","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 14:25:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"357e761bf7d4bafabe58967b752e40a3ef5bf853","revisions":{"f92e4bbc14b2d187ebe74b29da295b5ff3ecbdf2":{"kind":"REWORK","_number":1,"created":"2023-09-08 05:27:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/894252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/52/894252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/1"}}},"commit":{"parents":[{"commit":"a72267adf2c77e58a757dad360ba4ebff91b2496","subject":"Merge \"Revert \"Deprecate unused options of reserved store\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a72267adf2c77e58a757dad360ba4ebff91b2496"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-08 05:23:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-08 05:24:46.000000000","tz":540},"subject":"Fix legacy backend class usage with multi_store\u003dTrue","message":"Fix legacy backend class usage with multi_store\u003dTrue\n\nUsing legacy backend class (eg glance::backend::file) with multi_store\nset to True has been broken since [1], because of duplicate declaration\nof [glance_store] default_store option. This fixes the problem.\n\nChange-Id: Ibb12bdb8333fc91e2dc95c2c2cc4de79f2c82cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f92e4bbc14b2d187ebe74b29da295b5ff3ecbdf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/f92e4bbc14b2d187ebe74b29da295b5ff3ecbdf2"}]},"branch":"refs/heads/master"},"b002eab00916773c1ecc3e0ac23fa46b1b67cf94":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-08 05:33:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/894252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/52/894252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/2"}}},"commit":{"parents":[{"commit":"a72267adf2c77e58a757dad360ba4ebff91b2496","subject":"Merge \"Revert \"Deprecate unused options of reserved store\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a72267adf2c77e58a757dad360ba4ebff91b2496"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-08 05:23:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-08 05:32:49.000000000","tz":540},"subject":"Fix legacy backend class usage with multi_store\u003dfalse","message":"Fix legacy backend class usage with multi_store\u003dfalse\n\nUsing legacy backend class (eg glance::backend::file) with multi_store\nset to false(default) has been broken since [1], because of duplicate\ndeclaration of [glance_store] default_store option.\n\nThis fixes that problem.\n\nChange-Id: Ibb12bdb8333fc91e2dc95c2c2cc4de79f2c82cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b002eab00916773c1ecc3e0ac23fa46b1b67cf94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/b002eab00916773c1ecc3e0ac23fa46b1b67cf94"}]},"branch":"refs/heads/master"},"357e761bf7d4bafabe58967b752e40a3ef5bf853":{"kind":"REWORK","_number":3,"created":"2023-09-08 06:30:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/894252/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/52/894252/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/52/894252/3"}}},"commit":{"parents":[{"commit":"a72267adf2c77e58a757dad360ba4ebff91b2496","subject":"Merge \"Revert \"Deprecate unused options of reserved store\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a72267adf2c77e58a757dad360ba4ebff91b2496"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-08 05:23:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-08 06:30:24.000000000","tz":540},"subject":"Fix legacy backend class usage with multi_store\u003dfalse","message":"Fix legacy backend class usage with multi_store\u003dfalse\n\nUsing legacy backend class (eg glance::backend::file) with multi_store\nset to false(default) has been broken, because of duplicate declaration\nof [glance_store] default_store option.\n\nThis fixes that problem.\n\nChange-Id: Ibb12bdb8333fc91e2dc95c2c2cc4de79f2c82cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/357e761bf7d4bafabe58967b752e40a3ef5bf853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/357e761bf7d4bafabe58967b752e40a3ef5bf853"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
