)]}'
{"id":"openstack%2Fpuppet-glance~900353","triplet_id":"openstack%2Fpuppet-glance~stable%2Fwallaby~I576a58a846c232f599b6ba8c9e446bbc48582656","project":"openstack/puppet-glance","branch":"stable/wallaby","topic":"tripleo_wallaby_EOL","attention_set":{},"removed_from_attention_set":{"9976":{"account":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"last_update":"2023-11-24 16:13:58.000000000","reason":"Change was submitted"},"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2023-11-24 16:13:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I576a58a846c232f599b6ba8c9e446bbc48582656","subject":"Remove tripleo-based jobs ahead of Wallaby EOL","status":"MERGED","created":"2023-11-07 18:57:37.000000000","updated":"2023-11-24 16:13:58.000000000","submitted":"2023-11-24 16:13:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900353-tripleo_wallaby_EOL","meta_rev_id":"5d6fe3ffc0fb81d949f4a2f5704f90a0b8ec5d46","_number":900353,"virtual_id_number":900353,"owner":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-24 16:13:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-11-24 13:18:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-11-07 19:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-11-24 13:18:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-07 18:58:05.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2023-11-07 19:30:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-11-07 21:36:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-24 13:18:17.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"813a92aa99ec6a0c69e387eebfb7928363d6fe7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2023-11-07 18:57:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cd0a725f1c879697a02fec57ba4e443437f0ba7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2023-11-07 18:58:46.000000000","message":"Topic set to tripleo_wallaby_EOL","accounts_in_message":[],"_revision_number":1},{"id":"8c9a886edfd7d884334b88dd4e86f4b27ec83577","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-07 19:30:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b454e20c8369efac8dfaf4c530d83bb6462b7ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 21:36:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/222944d9bfeb417fb2fbc7594b7dcc36\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ccc180ab2534152b984652b3f315cd8 : FAILURE in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7cc055d95204bf29ac75154bdc795bc : SUCCESS in 5m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9657e9ca1f694a77aa2490be546c5209 : SUCCESS in 11m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa371eb2a01d473f898e55c62db38ff7 : FAILURE in 8m 58s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8f2080f8a394201a686f0f9001ee238 : FAILURE in 9m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4004262d8cd4b3cbf0688a9cb45d09d : FAILURE in 39m 06s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08ecbebd765f41bfac15e1b4dc9b4372 : FAILURE in 34m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6857ec413004662a934d31720276dab : FAILURE in 32m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd08468cf3c54100972be33e1906ee77 : FAILURE in 25m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/698391e167c94c038da5fce06cf30145 : TIMED_OUT in 2h 31m 24s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/49e06e205697440ea4b78df10851fed6 : SUCCESS in 1h 04m 12s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba0f007498114c4299956a4c4d40cf74 : TIMED_OUT in 2h 31m 20s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3faf9a90b9274e3fa27290d0794eb562 : TIMED_OUT in 2h 31m 15s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2ac050c345345b3a865348350e7b0df : SUCCESS in 29m 04s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d4569256b8e47cfbdad17cc92861a2e : SUCCESS in 14m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1206b8071f1259578b52f4d5b6079875870b9da7","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2023-11-08 13:49:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24f88e4aed8cb8a2686457ef4d8d96e8e2c1c869","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-24 13:18:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c662da6221c2a35c28b25f0fb390b27c0e8d827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 14:49:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab45218b1f444058b27e3d1617a46a36\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7babc2ee3efa4c1e953ce5de112c0010 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d188e737a54b4ba590082a4123df7a76 : SUCCESS in 6m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1502af83369b40f9b8e17d8f9b44510b : SUCCESS in 14m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e144b7d88774b7a9de0459ffb260e91 : SUCCESS in 9m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5a1a9b2d59449c8a7a95d291d75d18a : SUCCESS in 9m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42677b6cda3f4b85a04c0a52023db843 : FAILURE in 42m 02s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da7cc9092cf9412ea18fa97506be68e0 : FAILURE in 37m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc805bb744194de99a65a4a2a838846c : FAILURE in 32m 34s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c75b35d701ce438685f49054899a3893 : FAILURE in 25m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c9ee8fded5b46d197e3d7b2146b12d5 : SUCCESS in 1h 16m 10s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0db25b3bea94ca992817f5f5b56e3f8 : SUCCESS in 1h 23m 07s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/58b68422f34042aaa96f4dfa80464012 : SUCCESS in 1h 22m 51s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7eb598da6f714dc9845abef1810c5290 : SUCCESS in 53m 52s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd154853898e4f02abeb0527f7536c1b : SUCCESS in 25m 34s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c2160aeb2ff4155b9a0658d8f2516f3 : SUCCESS in 17m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b94a58fe7460629f8b178fc1670c1d7c6bcafcb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 14:50:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8662cda25fca908c78ffa9a10b54858b0c4d1eb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 16:13:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3011d81cd074ff6827baa856a20853e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08962eeffc574ed4a20e6fa1394f8109 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43eaa86e3168443d9dbc5dadf8b197d6 : SUCCESS in 4m 49s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d77b86dbcfb74e2ba694b01befd69319 : SUCCESS in 13m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b97940b150244f779660f103e9c94b1e : SUCCESS in 7m 55s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/65a128128bd44fcea0258c22a81f3c65 : SUCCESS in 1h 18m 48s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/49d0868362ee4310b5c00d5919340bc0 : SUCCESS in 1h 08m 42s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ccfe82780aa4a15abb811310f1a751b : SUCCESS in 1h 13m 04s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0710c78e25f843b58a821362e76fa2f4 : SUCCESS in 56m 03s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a48008e39a7409ca65cf79ea78960b7 : SUCCESS in 23m 12s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8d9c2e3e3f84ec18fdac38202a6c422 : SUCCESS in 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d6fe3ffc0fb81d949f4a2f5704f90a0b8ec5d46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 16:13:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"255ec9b3718f0def74974ad9659ddc58b8215694","revisions":{"255ec9b3718f0def74974ad9659ddc58b8215694":{"kind":"REWORK","_number":1,"created":"2023-11-07 18:57:37.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/53/900353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/53/900353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/53/900353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/53/900353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/53/900353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/53/900353/1"}}},"commit":{"parents":[{"commit":"81b081d759883b23fa01d5ef6abdd3c0c41b603d","subject":"Add support for MultiStrOpt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/81b081d759883b23fa01d5ef6abdd3c0c41b603d"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2023-11-07 18:55:17.000000000","tz":-300},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2023-11-07 18:57:37.000000000","tz":-300},"subject":"Remove tripleo-based jobs ahead of Wallaby EOL","message":"Remove tripleo-based jobs ahead of Wallaby EOL\n\nWallaby branches on TripleO based repos\nwill be going EOL soon.\nThis review removes the TripleO based jobs\nto avoid possible future failures.\n\nChange-Id: I576a58a846c232f599b6ba8c9e446bbc48582656\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/255ec9b3718f0def74974ad9659ddc58b8215694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/255ec9b3718f0def74974ad9659ddc58b8215694"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
