)]}'
{"id":"openstack%2Fpuppet-glance~901104","triplet_id":"openstack%2Fpuppet-glance~master~I57a2ecc763ccac52d9122afa86871ad74bca7450","project":"openstack/puppet-glance","branch":"master","topic":"provider-cleanup","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-17 19:05:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57a2ecc763ccac52d9122afa86871ad74bca7450","subject":"Stop reading keystone_authtoken options","status":"MERGED","created":"2023-11-16 06:11:03.000000000","updated":"2023-11-17 19:06:32.000000000","submitted":"2023-11-17 19:05:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901104-provider-cleanup","meta_rev_id":"039299e9ed22c8a0b3268fa1090c677e089c8e3c","_number":901104,"virtual_id_number":901104,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 19:05:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-11-17 14:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-11-17 15:10:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-11-17 15:10:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 07:35:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-17 14:05:54.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-11-17 15:10:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"d08c684459a6f6986732c03b6787394e47c83a09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:11:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b19de3882a4a572d167c70f1771c6788faa3e71","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:11:21.000000000","message":"Topic set to provider-cleanup","accounts_in_message":[],"_revision_number":1},{"id":"777383aa100b223b6ef27c56bfe21af68f89a7d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:16:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e668248a732935454d17f34076eb0beda245e7dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:19:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a7bd1abf6e992963223746fc3107de23e6c55ec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:26:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7a6e9ad333c7256286601127ebf7730db342323f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 07:35:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7aa9f1cb2ec44d3bab7b3bb55dda6c5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/470e8ddb289e4bb1a3ea5e12d3610c21 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e27677fc08c46d2abed4c1ee7f757be : SUCCESS in 2m 43s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b2df1327d504832954b1676fa8b4c4a : SUCCESS in 5m 40s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ae7e867fa8a4424ab49770b6aefb15f : SUCCESS in 12m 09s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9305e35a1dca4f61b0ba42d16a94d405 : SUCCESS in 7m 34s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/036b429d66954f7782d83b205c3004d2 : SUCCESS in 10m 12s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/905c037431f74f1ca5510481f26b4433 : SUCCESS in 7m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/381dc189722745f8884692b217515822 : SUCCESS in 41m 30s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c21839d83ab24bd1bfe62e4b835a0e59 : SUCCESS in 1h 07m 40s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce0770c93bdf437cb9eeafd789b854a3 : SUCCESS in 49m 52s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6d1abb547ae489e9870e56b83dc0d79 : SUCCESS in 38m 21s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f229fc49d2a4bbda756e12c22dd2dc4 : SUCCESS in 45m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5c84511277f46bdac5375f50d7b388d : SUCCESS in 51m 15s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec1d77790876451898e5fe85e90e5d6d : SUCCESS in 1h 06m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49f831c3d4f2416a9efb233ad4a5f7bc : SUCCESS in 51m 44s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/987fbb4c18294eca9e1fcb1c484b9ce9 : SUCCESS in 57m 58s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54436df6bd454d709b329d087d890408 : SUCCESS in 44m 59s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9fb8e495da54c1eb66128c5e8e503e5 : SUCCESS in 16m 23s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efa11e15f5ec402eb788d6684b6f4102 : SUCCESS in 11m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76f02335dd3648a1a3f1e4929ff8fbb3 : SUCCESS in 12m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f670afd325994dfa85c0b36592aab790 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":4},{"id":"59abffffa48baf57ef905f51b0fb40154b887322","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-11-17 14:05:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d4f6a712e06dafbf63d1bf70643f2921be3fcbb0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-17 15:10:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4d064725582f4904e1a523df57605fc348760e56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 15:11:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cefbce6e6c6ceaf330af265db6fecc8a58d21713","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:05:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f246a98180204990be1cc6938084197b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93729aea5040491287ccc633e2607d16 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57cf2de25d8642ec9558660d1547c768 : SUCCESS in 2m 59s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1154bf79869402e9bd47209a207fc6b : SUCCESS in 12m 55s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fa5c4104d5a42e4a50dfdfc6379d933 : SUCCESS in 11m 23s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/04c28be0a1cd4340b8ddb5720dfa5f40 : SUCCESS in 11m 21s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/038e32250f2b481c875cfb0f152220d4 : SUCCESS in 1h 02m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9d7beb049ba4f9587fc7c2cd0a935c8 : SUCCESS in 1h 26m 35s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/899e0f91f68d4d7bb30e967bd8595e5b : SUCCESS in 1h 15m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5762b92cc8934a8d83ca4a12f876431c : SUCCESS in 59m 43s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c805e6675ad4758adeb20ffaf49b45b : SUCCESS in 58m 53s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/34022042785c4392a50e257c902ffb09 : SUCCESS in 19m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5f91de53e524b3287ff364f88323a49 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":4},{"id":"14d141073c293fc64db1d4e7d32b9d90a4d10ff7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:05:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"039299e9ed22c8a0b3268fa1090c677e089c8e3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:06:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de5a4fd9fee64ed998223575f49cae4d\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5901ac76d88f4fc48a5406e1d145f8e6 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"50e67be163244f128b1ff08d3b051b0539fc16b7","revisions":{"5129e3a92bb0181ad60ad50962d4b31910751942":{"kind":"REWORK","_number":1,"created":"2023-11-16 06:11:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/901104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/04/901104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/1"}}},"commit":{"parents":[{"commit":"1317af56d4f4700e4300024e5344c8b0ac780eeb","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1317af56d4f4700e4300024e5344c8b0ac780eeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:10:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:10:30.000000000","tz":540},"subject":"Stop reading keystone_authtoken options","message":"Stop reading keystone_authtoken options\n\nUsing credentials in keystone_authtoken options for glance_* resources\nwas deprecated some cycles ago[1].\n\n[1] 17cc71c509b6b35a3cc1474c854f41d37cf6bba2\n\nChange-Id: I57a2ecc763ccac52d9122afa86871ad74bca7450\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5129e3a92bb0181ad60ad50962d4b31910751942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/5129e3a92bb0181ad60ad50962d4b31910751942"}]},"branch":"refs/heads/master"},"fa074cbc103031a420a0a8fdbd5f30b3a50f8ca2":{"kind":"REWORK","_number":2,"created":"2023-11-16 06:16:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/901104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/04/901104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/2"}}},"commit":{"parents":[{"commit":"1317af56d4f4700e4300024e5344c8b0ac780eeb","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1317af56d4f4700e4300024e5344c8b0ac780eeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:10:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:16:30.000000000","tz":540},"subject":"Stop reading keystone_authtoken options","message":"Stop reading keystone_authtoken options\n\nUsing credentials in keystone_authtoken options for glance_* resources\nwas deprecated some cycles ago[1].\n\n[1] 17cc71c509b6b35a3cc1474c854f41d37cf6bba2\n\nChange-Id: I57a2ecc763ccac52d9122afa86871ad74bca7450\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fa074cbc103031a420a0a8fdbd5f30b3a50f8ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fa074cbc103031a420a0a8fdbd5f30b3a50f8ca2"}]},"branch":"refs/heads/master"},"ecd3ef39dbc99482b9017dc1fe764525526cae6b":{"kind":"REWORK","_number":3,"created":"2023-11-16 06:19:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/901104/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/04/901104/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/3"}}},"commit":{"parents":[{"commit":"1317af56d4f4700e4300024e5344c8b0ac780eeb","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1317af56d4f4700e4300024e5344c8b0ac780eeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:10:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:19:10.000000000","tz":540},"subject":"Stop reading keystone_authtoken options","message":"Stop reading keystone_authtoken options\n\nUsing credentials in keystone_authtoken options for glance_* resources\nwas deprecated some cycles ago[1].\n\n[1] 17cc71c509b6b35a3cc1474c854f41d37cf6bba2\n\nChange-Id: I57a2ecc763ccac52d9122afa86871ad74bca7450\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ecd3ef39dbc99482b9017dc1fe764525526cae6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ecd3ef39dbc99482b9017dc1fe764525526cae6b"}]},"branch":"refs/heads/master"},"50e67be163244f128b1ff08d3b051b0539fc16b7":{"kind":"REWORK","_number":4,"created":"2023-11-16 06:26:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/901104/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/04/901104/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/04/901104/4"}}},"commit":{"parents":[{"commit":"1317af56d4f4700e4300024e5344c8b0ac780eeb","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1317af56d4f4700e4300024e5344c8b0ac780eeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:10:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:26:30.000000000","tz":540},"subject":"Stop reading keystone_authtoken options","message":"Stop reading keystone_authtoken options\n\nUsing credentials in keystone_authtoken options for glance_* resources\nwas deprecated some cycles ago[1].\n\n[1] 17cc71c509b6b35a3cc1474c854f41d37cf6bba2\n\nChange-Id: I57a2ecc763ccac52d9122afa86871ad74bca7450\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/50e67be163244f128b1ff08d3b051b0539fc16b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/50e67be163244f128b1ff08d3b051b0539fc16b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
