)]}'
{"id":"openstack%2Fpuppet-glance~916748","triplet_id":"openstack%2Fpuppet-glance~master~I5fed59ee57e289d7b979163725ee6c676858be5c","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-23 14:26:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5fed59ee57e289d7b979163725ee6c676858be5c","subject":"Remove unnecessary openstacklib::policy resource tag","status":"MERGED","created":"2024-04-23 09:15:16.000000000","updated":"2024-04-23 14:26:19.000000000","submitted":"2024-04-23 14:26:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916748","meta_rev_id":"1cd108c89b6377b1464716abce4cdae8498312c3","_number":916748,"virtual_id_number":916748,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-23 14:26:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2024-04-23 12:40:13.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2024-04-23 13:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2024-04-23 13:08:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-23 10:43:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-23 12:40:13.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2024-04-23 13:08:57.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b6f457c6e8652206ca42f3354509ff5c22126b17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-23 09:15:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"349003c34e92e500f5329edf1fb6fef9b82a9c60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 10:43:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee5eea4395bf48aba26351c041d155ca\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee4cdcabee34428cbb9a98e660328b02 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b24c82d4d1343d395906066f960758f : SUCCESS in 2m 44s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/637a64b8ea3647a6a6ef86a5e01f5190 : SUCCESS in 5m 40s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/550ddb04319b4a3eb00c9941ccb1ebae : SUCCESS in 10m 22s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6789eecfaa014b80a1b6661c783f1fe8 : SUCCESS in 9m 17s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/372d94cb229740cc952feaa9d6dc051d : SUCCESS in 10m 11s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2b1d0f58cfa4c00aecc174aa9d98986 : SUCCESS in 8m 06s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/551893fba7064f52856cebe9d87949fa : SUCCESS in 46m 33s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1034bb136f24a48adb9cce99a94e0ec : FAILURE in 39m 32s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8102276c8be49ac9ff0675f986e9671 : FAILURE in 33m 12s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2231a25567a456a9207268359091e41 : SUCCESS in 38m 54s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae0db84089324b84b755b094a002450d : SUCCESS in 43m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/afe87d67042e417ab320373525d81921 : SUCCESS in 58m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f18da7cf9eb48cf995029bf2ef98e0b : SUCCESS in 1h 02m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ec4d7cf11ce456c9c68169926b60324 : SUCCESS in 56m 52s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4476181e1cf44bdb23b34092118e96c : SUCCESS in 54m 36s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/04041a0022874942a216ae8bad1a7d1f : SUCCESS in 46m 47s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/b41f3e85dbf446128ed19c3f2077f5e9 : SUCCESS in 20m 34s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24c67d53a06a4d538a92372b9cfae8fc : SUCCESS in 13m 58s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba390e7601784de49604b0bff97b77c6 : SUCCESS in 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e405dc88b4fd78365077ec7dcd8e82c87c4fe017","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2024-04-23 12:40:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"005ac810daf594331932273d2544c47e090f979d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-04-23 13:08:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e75cccd04dcb1f5590ff6cce40c0499b9ecd628","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 13:09:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5469b87800e7bc71d66fdc8af754bd108f881482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 14:26:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef9037803cd54d5aba418718a214c6db\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a109b5c634c452597bf77f4d077fa26 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05cd68637bc249dab9c888fe6714d45f : SUCCESS in 3m 08s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/e69f45c343cb4c23bbbc36fcc6d26cb5 : SUCCESS in 11m 43s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0be367ac093f4d678c754fec978bf113 : SUCCESS in 8m 55s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/973aa8c36a224d3a865f43f1445f0cbd : SUCCESS in 58m 28s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdb175f21096456b8584293d42ec03a8 : SUCCESS in 1h 08m 06s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ed602d5f5874895944eb9bed2191a97 : SUCCESS in 53m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/368550bef77b4421875305ffd1ab0be5 : SUCCESS in 59m 05s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/461d704230b8498fafbf36f6d5728529 : SUCCESS in 52m 53s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/e48318474a0449d88581218b92482d80 : SUCCESS in 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1cd108c89b6377b1464716abce4cdae8498312c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 14:26:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"962b7474b0043dd969523ef86d7cc8c48d0755cd","revisions":{"962b7474b0043dd969523ef86d7cc8c48d0755cd":{"kind":"REWORK","_number":1,"created":"2024-04-23 09:15:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/916748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/48/916748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/48/916748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/48/916748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/48/916748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/48/916748/1"}}},"commit":{"parents":[{"commit":"9232855977c8a37225962f3fa0955374d7af268a","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9232855977c8a37225962f3fa0955374d7af268a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-23 09:14:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-23 09:14:49.000000000","tz":540},"subject":"Remove unnecessary openstacklib::policy resource tag","message":"Remove unnecessary openstacklib::policy resource tag\n\nThis tag is no longer needed since dependency refactoring[1]. Remove\nthe tag to make the implementation consistent across modules.\n\n[1] 8b3054a13f6af070d627a019c9e1fd996fd1fd6a\n\nChange-Id: I5fed59ee57e289d7b979163725ee6c676858be5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/962b7474b0043dd969523ef86d7cc8c48d0755cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/962b7474b0043dd969523ef86d7cc8c48d0755cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
