)]}'
{"id":"openstack%2Fpuppet-glance~924705","triplet_id":"openstack%2Fpuppet-glance~master~Ia3c7ac7025f0a0f908fed11feb229d159d72f890","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-23 15:55:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3c7ac7025f0a0f908fed11feb229d159d72f890","subject":"Add support for [healthcheck] enable_by_file_paths","status":"MERGED","created":"2024-07-23 10:10:48.000000000","updated":"2024-07-23 15:57:02.000000000","submitted":"2024-07-23 15:55:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924705","meta_rev_id":"679c09266b4ccbcff7a397cc72cdc47889737171","_number":924705,"virtual_id_number":924705,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-23 15:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2024-07-23 12:29:53.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2024-07-23 14:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2024-07-23 14:41:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 11:37:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 12:29:53.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2024-07-23 14:41:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e5027a43daf800acf1c028e601de0b72183ea0ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 10:10:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"589f00f8ab4c94aa45c6afbfd657726231738a0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 11:37:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa6b6776e7e46f5b1442ef243bb6a05\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84addf25659e4a1bb84c2b82d865c1f0 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97f0b46782af4aab94f170974e09e643 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b09ab267cc7546238c000ee2619449ac : SUCCESS in 6m 23s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bcc6b0fca1847b1b286c6e7078cbde2 : SUCCESS in 11m 02s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b46530c6aea24bf7b8bf11a3285c085f : SUCCESS in 7m 52s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ddfc0442d104aaa99a83b0a2bb894cd : SUCCESS in 10m 11s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9a87d88339f4c9db6540172b8a8025c : SUCCESS in 8m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9844e4cd0f39436b95476df3822f175d : SUCCESS in 50m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/874018fcb8ef4d8fa7f602ddb6275025 : SUCCESS in 57m 08s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0613f4bf8c25433aa505d385ef783840 : SUCCESS in 46m 57s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d03e89be8de4f99a7943f134d57612a : SUCCESS in 46m 50s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5712541273549459d9c82143219de97 : SUCCESS in 40m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1211bdbb76204b75b9f85541846622a8 : SUCCESS in 51m 47s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/444ce8233c5d4ae8a73bac2a26efdac6 : SUCCESS in 1h 24m 52s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f321f15e6c044ea2a967a1d9d385da7f : SUCCESS in 1h 04m 32s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/536e98bfcc0c42f2a753624755949683 : SUCCESS in 1h 09m 29s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3b11d65475c41adad80d07829d3d754 : SUCCESS in 56m 24s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d6d31a9335841eb8cc2f054333f73af : SUCCESS in 19m 31s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33c7bc823f8d4fe2ace2dc92aa9cb378 : SUCCESS in 11m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e0ea73469f84e239a70155b5792cc13 : SUCCESS in 13m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/228405b684744bdb8d1cc6a21edaf3f1 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"86861693d65e799e7d8cb6531ef22812d260cf29","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2024-07-23 12:29:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"48066f1954afffc11be31da645dc07b318ef86e0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-07-23 14:41:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7e528c4141347ec588b956f5c3e87a29ddff03f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 14:42:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d53ad58ec5c6cd6316795d3fda5e599c8bdc298a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 15:55:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7391b970bd03493fa235c3e35d7934bd\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/770ec5c903a447c8bc105e7d4976b645 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a2b269b8bf8441b8ec4a6f915a815f3 : SUCCESS in 2m 34s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/44e59a0c87ae499b857de4754cb0a780 : SUCCESS in 10m 27s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a97fe1c763c54b758724bd4728f83140 : SUCCESS in 8m 19s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8558f80d737418e8cb720abc02f0e69 : SUCCESS in 53m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7d19099d2444e48a1de2381b293032d : SUCCESS in 1h 12m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd83eaa6061b4744ac159447ed5de118 : SUCCESS in 57m 20s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2d494dc1a7644b99ffd37262115846e : SUCCESS in 59m 22s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0db607d180fc47adaa9959d3495cd15a : SUCCESS in 47m 35s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1d0239f6a9e458fb6cfd1accec57bc0 : SUCCESS in 18m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/690b2c3a0e2c46308b0323229610389e : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"abb191ff6e5d0340a7ae23cab71ae3d1baa7726b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 15:55:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"679c09266b4ccbcff7a397cc72cdc47889737171","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 15:57:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e64f3ea1927a471fa9181108db466aeb\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd1b03aa8c734d2b8dcf70be93413785 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24c2d603bc46edd17d48a2ea08b0b1d5773c3cd5","revisions":{"24c2d603bc46edd17d48a2ea08b0b1d5773c3cd5":{"kind":"REWORK","_number":1,"created":"2024-07-23 10:10:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/924705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/05/924705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/924705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/924705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/924705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/05/924705/1"}}},"commit":{"parents":[{"commit":"8cc7c22dde94342325211f5c4b29126e532f09a4","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8cc7c22dde94342325211f5c4b29126e532f09a4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 10:10:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 10:10:30.000000000","tz":540},"subject":"Add support for [healthcheck] enable_by_file_paths","message":"Add support for [healthcheck] enable_by_file_paths\n\nDepends-on: https://review.opendev.org/923648\nChange-Id: Ia3c7ac7025f0a0f908fed11feb229d159d72f890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/24c2d603bc46edd17d48a2ea08b0b1d5773c3cd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/24c2d603bc46edd17d48a2ea08b0b1d5773c3cd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
