)]}'
{"id":"openstack%2Fpuppet-glance~930133","triplet_id":"openstack%2Fpuppet-glance~master~Ieb3480254411f1251df7bbcdc85b3952b46f3335","project":"openstack/puppet-glance","branch":"master","topic":"watch_log_file","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-23 16:22:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ieb3480254411f1251df7bbcdc85b3952b46f3335","subject":"logging: Deprecate watch_log_file","status":"MERGED","created":"2024-09-21 16:55:13.000000000","updated":"2024-09-23 19:02:26.000000000","submitted":"2024-09-23 19:01:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930133-watch_log_file","meta_rev_id":"da3c71ebc158ce1b48915904183cc9a595899281","_number":930133,"virtual_id_number":930133,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-23 19:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2024-09-23 03:48:19.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2024-09-23 14:44:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2024-09-23 14:44:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-21 17:23:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-23 03:48:19.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2024-09-23 14:44:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"86df355f4b0d5243e5b46aaa6f76e33f50601ae8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-21 16:55:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5cc3bf9149985facad91c620ede16e0763370a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-21 17:19:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"95dcd88a8bfdfb4d44a640c3232c37171f606b77","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-21 17:22:46.000000000","message":"Topic set to watch_log_file","accounts_in_message":[],"_revision_number":2},{"id":"780247a3fe5a0815db917cbe3bdecf1c21a5cf0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 17:23:05.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/930126 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"1fb16c0aa8b40c370deaf1c365d035198bd16e89","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-21 17:24:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"933d64e44118e5f44bb044150df2447351c23150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-21 17:26:42.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3723545b23d4a5cadd70cc4f3a5bfd8623c56cd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 17:31:50.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/930126 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"b9799211dfbd7be7ede6d05cdbbdd0a2882feb65","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-21 18:11:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b4057ee6dfd164298bd330ae0c0fd49f9904e949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 19:35:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffbe79bf7d5a4ed2b7301f19e7e1033f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/669907660b6a4c9ca258e1d2a6e310b2 : SUCCESS in 2m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/220f87cb53054d5f99452f4f2c42615a : SUCCESS in 2m 28s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e258c0c3e22d40a3b59cd369dc05b04a : SUCCESS in 3m 43s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3f939b80313463b8b3a4655625e2f30 : SUCCESS in 8m 37s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c17360e0e7541b8afa950aeac636e08 : SUCCESS in 4m 39s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/75f3063215404b8e964ef5f68cd6a484 : SUCCESS in 8m 00s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a40f77dca2d4425ea51a3ee4c5547e38 : SUCCESS in 6m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8da6b9df81184b96a423295723f00ad9 : SUCCESS in 27m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c7a425d93714fdebd1e61af005cc428 : SUCCESS in 49m 54s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66bff632d5b843ad881182ccdc8b954d : SUCCESS in 44m 46s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73edfe40434a42eeb4611de6d0a7ef89 : SUCCESS in 47m 22s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/836eafb556a14840adda98ef83ce7b67 : SUCCESS in 40m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/22dc43ea64b34a28b9ce5b53d0390b15 : SUCCESS in 53m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea4655d0d94843dc960bb11dac5b88f9 : SUCCESS in 1h 23m 13s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f8fdc800910420780a407b1bd154988 : SUCCESS in 49m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aedc8e7dc67941128d53c0ca9ba4dc32 : SUCCESS in 55m 35s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c37aa53c954e4ebd885ecedc0387b4d1 : SUCCESS in 50m 14s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/91eca91248504853808503a7cabe31a0 : SUCCESS in 17m 26s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8da73852d5564d879988fad05c80a377 : SUCCESS in 8m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5722df8b3bd4ef9af6a437a92cfba68 : SUCCESS in 11m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30b734f81ca845dfa6afe58c1b310aac : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d1de87ae78f270545055b49dbb7bc9947879d1fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-22 03:30:38.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"addc5efe51c0b307cedeb9052bbf2a926e7666d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-22 04:50:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d053f7c2386546b0919496666ad8d96f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c08eccd002f4c75a3dd1a1a10457c26 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0625c9b79534d4f902b242eeea0bfa9 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed84bf3f4e8a4eebb1878bd1c015abf8 : SUCCESS in 6m 40s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/13b242fec00549d7af3d037fcbe80f8b : SUCCESS in 10m 45s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ec235a54dac4d2db96cbeb2a1189f40 : SUCCESS in 7m 59s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/cea5ad50013c40f4a7436eec75d46271 : SUCCESS in 9m 28s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca07d20f730342ed892a43c9e5b99970 : SUCCESS in 7m 06s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b54a8c4ea31490cae5e7b3d831259f0 : SUCCESS in 39m 24s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa3b30f8e6cb40d09d2c03a734b1417a : FAILURE in 30m 41s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b322e0282fe24ed5a1d6d949e82ef243 : SUCCESS in 44m 05s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8afeabf8b0744ee382b177986e9061e9 : SUCCESS in 48m 07s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8eee9e2064574d64a096766a67065472 : SUCCESS in 36m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9df10f0b1a914f0a999e93d49de5c87d : SUCCESS in 54m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b7b2a1a5d10494d92eef5efb2e40369 : SUCCESS in 1h 14m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/760b4467c71d4bf7b3de9f9830af6bc2 : SUCCESS in 36m 11s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5cea288d7124239b7efc647f4f797f0 : SUCCESS in 43m 05s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c5311433abe470ea2febd81702f7924 : SUCCESS in 42m 31s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa3dd892f5c84767bc6c4a58ef189753 : SUCCESS in 17m 47s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef54921989b54be09cff946264a4dd48 : SUCCESS in 11m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6879ecb31cf54ca1ac760a4b341c0df2 : SUCCESS in 13m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa4fc81474574a608907269b30e6c80b : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":4},{"id":"7b6f501bb5f8945e8da6027748bf48cb08a5d1cd","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2024-09-23 03:48:19.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"26b93908026351eb1b48218343df71ea48046a1d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-09-23 14:44:31.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b0c7d4484db4ac1d40e6b8c93974b9a78ac54b72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 14:44:42.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc9fd27a021e4f00a8df8ade40dbd51f\n\n\nWarning:\n  Change 930126 in project openstack/puppet-oslo does not share a change queue with 930133 in project openstack/puppet-glance","accounts_in_message":[],"_revision_number":4},{"id":"d0136b3025e81f0e745134e6c2c636ec6eac891b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-23 16:22:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c77195eb22ed456594a537e72672341a5ad08d07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 17:45:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a660c0ebe53f41f3b035ccd069435dd1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2293548fc3543a8beb3ea2b3453cd45 : SUCCESS in 2m 28s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/937b011621c341289cd3719ff492d161 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/34e0d758b23641b398cc1305d9cbcf67 : SUCCESS in 4m 57s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3577910f1034405393f727a2987eff0f : SUCCESS in 9m 05s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b1f63c819104593adf95bf1e7803f1c : SUCCESS in 6m 19s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc36a608db5443db95b7b32411665c57 : SUCCESS in 7m 23s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97e183e1ef834a67bf932f8dcfe0e52f : SUCCESS in 6m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c230a4e36e5048b9868eb45d2a01f88e : SUCCESS in 43m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73529760dc8f4f22aff6e0e7117fb615 : SUCCESS in 56m 36s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cafc5da27c349da90fb7dd2e2a5b366 : SUCCESS in 45m 17s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99046cc1ab0d4b6dbf48df0c023b742c : SUCCESS in 41m 29s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/682ea958ddc840c2baf2c21850f5cf90 : SUCCESS in 39m 49s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab52469e168b49e3817a049f01222577 : SUCCESS in 53m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/267d28b71fb3491291e841b4551b7834 : SUCCESS in 1h 18m 43s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7da83d3769f14d8c95dd3fb7944db4ba : SUCCESS in 35m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a761779ffd214b21ab38f7b7ebb35c4c : SUCCESS in 54m 17s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff6c2bb1abf849a5a22fbc474c7ffdb8 : SUCCESS in 43m 41s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f87bfc998c504064ab6c2d8bd315b3dd : SUCCESS in 13m 56s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2fedb565da24b4fa7b11b85752ffb3d : SUCCESS in 13m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9cac4a3f42c4cf69413abc3bc464085 : SUCCESS in 12m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f1f1ac405934d098d165a3a68a1a487 : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":4},{"id":"a9969ed680282bfffeb610ed2f3e6be677330da5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 17:45:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b878d16e085a56a2e561a1464f126d001f6bfd3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 19:01:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ce2966018fb466facd34dd724b3f63d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a9ad6e62dd844f8acb8f53f765289e8 : SUCCESS in 2m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f106acf45ec46a983313e55487137b2 : SUCCESS in 2m 32s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/4726a4218b2544abad199e9540f425db : SUCCESS in 8m 08s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6134497c77cc4e71a5d791aa00f9d78c : SUCCESS in 6m 11s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7f0132ac0a5478fa6a54fc092bff660 : SUCCESS in 52m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61e50e08c4e74ec7984195a53a63f77b : SUCCESS in 1h 14m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/40e506f54fc148e6bf6dcceb254eb763 : SUCCESS in 49m 43s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/57b4486e128b4495a7370b5367c6269b : SUCCESS in 53m 11s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbaf717eaffc4500ad1cf280d870ed86 : SUCCESS in 47m 34s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3370c1d3c9e44a2a98a2f8fa6e90b008 : SUCCESS in 18m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed90eabe2104467bbd1c5d21ae397725 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":4},{"id":"1c180e413bb8e39e9e94fe0f3f61f719f976a3fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 19:01:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"da3c71ebc158ce1b48915904183cc9a595899281","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 19:02:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1518071444f045e887c5c9f550a4b869\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc2efb0c6f494fbebe1cade75a9d4076 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c29243ded4042df292deb1d1d0488a2e6524fb38","revisions":{"54e1577721d23442137c3dab2b251e39370ea5e4":{"kind":"REWORK","_number":1,"created":"2024-09-21 16:55:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/930133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/33/930133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/1"}}},"commit":{"parents":[{"commit":"711abd750a1f603d382a7b9b396b096ff9d61bd7","subject":"Add validations for keystone resource parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/711abd750a1f603d382a7b9b396b096ff9d61bd7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-21 16:55:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-21 16:55:01.000000000","tz":540},"subject":"logging: Deprecate watch_log_file","message":"logging: Deprecate watch_log_file\n\nThis option was deprecated in oslo.log 6.0.0 [1]. Because the warning\nto catch usage of the parameter is implemented in the underlying\noslo::logging defined resource type, no additional warning message is\nimplemented in this module.\n\nDepends-on: https://review.opendev.org/930126\nChange-Id: Ieb3480254411f1251df7bbcdc85b3952b46f3335\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/54e1577721d23442137c3dab2b251e39370ea5e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/54e1577721d23442137c3dab2b251e39370ea5e4"}]},"branch":"refs/heads/master"},"31ae8638f618481944d9c45671712e13ba74b9b4":{"kind":"REWORK","_number":2,"created":"2024-09-21 17:19:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/930133/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/33/930133/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/2"}}},"commit":{"parents":[{"commit":"711abd750a1f603d382a7b9b396b096ff9d61bd7","subject":"Add validations for keystone resource parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/711abd750a1f603d382a7b9b396b096ff9d61bd7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-21 16:55:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-21 17:19:03.000000000","tz":540},"subject":"logging: Deprecate watch_log_file","message":"logging: Deprecate watch_log_file\n\nThis option was deprecated in oslo.log 6.0.0 [1]. Because the warning\nto catch usage of the parameter is implemented in the underlying\noslo::logging defined resource type, no additional warning message is\nimplemented in this module.\n\nDepends-on: https://review.opendev.org/930126\nChange-Id: Ieb3480254411f1251df7bbcdc85b3952b46f3335\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/31ae8638f618481944d9c45671712e13ba74b9b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/31ae8638f618481944d9c45671712e13ba74b9b4"}]},"branch":"refs/heads/master"},"ead5eaf3199a297979fbf9669288d2e073e53f26":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-09-21 17:26:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/930133/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/33/930133/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/3"}}},"commit":{"parents":[{"commit":"711abd750a1f603d382a7b9b396b096ff9d61bd7","subject":"Add validations for keystone resource parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/711abd750a1f603d382a7b9b396b096ff9d61bd7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-21 16:55:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-21 17:26:42.000000000","tz":0},"subject":"logging: Deprecate watch_log_file","message":"logging: Deprecate watch_log_file\n\nThis option was deprecated in oslo.log 6.0.0 [1]. Because the warning\nto catch usage of the parameter is implemented in the underlying\noslo::log defined resource type, no additional warning message is\nimplemented in this module.\n\nDepends-on: https://review.opendev.org/930126\nChange-Id: Ieb3480254411f1251df7bbcdc85b3952b46f3335\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ead5eaf3199a297979fbf9669288d2e073e53f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/ead5eaf3199a297979fbf9669288d2e073e53f26"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c29243ded4042df292deb1d1d0488a2e6524fb38":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-09-22 03:30:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/930133/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/33/930133/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/33/930133/4"}}},"commit":{"parents":[{"commit":"711abd750a1f603d382a7b9b396b096ff9d61bd7","subject":"Add validations for keystone resource parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/711abd750a1f603d382a7b9b396b096ff9d61bd7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-21 16:55:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-22 03:30:38.000000000","tz":0},"subject":"logging: Deprecate watch_log_file","message":"logging: Deprecate watch_log_file\n\nThis option was deprecated in oslo.log 6.0.0 [1]. Because the warning\nto catch usage of the parameter is implemented in the underlying\noslo::log defined resource type, no additional warning message is\nimplemented in this module.\n\n[1] https://review.opendev.org/914788\n\nDepends-on: https://review.opendev.org/930126\nChange-Id: Ieb3480254411f1251df7bbcdc85b3952b46f3335\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c29243ded4042df292deb1d1d0488a2e6524fb38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c29243ded4042df292deb1d1d0488a2e6524fb38"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
