)]}'
{"id":"openstack%2Fpuppet-glance~931744","triplet_id":"openstack%2Fpuppet-glance~master~I69dcb3bb3b479248ff5e2e84ec93a4b29b9dba22","project":"openstack/puppet-glance","branch":"master","topic":"parsepythondict","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-09 15:27:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69dcb3bb3b479248ff5e2e84ec93a4b29b9dba22","subject":"Use common function to parse python dict","status":"MERGED","created":"2024-10-08 08:08:55.000000000","updated":"2024-10-09 15:27:56.000000000","submitted":"2024-10-09 15:27:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931744-parsepythondict","meta_rev_id":"20e3b94deb15b2409a2e2cae876240b6907dd3be","_number":931744,"virtual_id_number":931744,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-09 15:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-10-09 07:36:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-10-09 14:15:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-10-09 14:15:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-08 08:25:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-09 07:36:18.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-10-09 14:15:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"74be09de7c6810e2f5f24cb3f77e081f3b7ff07f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-08 08:08:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"238a4eda92267c2b0e13e09111d45efe3a6447fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-08 08:18:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eca571c0d43f68187fd9ab002e6dc9be70ae0757","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-08 08:19:17.000000000","message":"Topic set to parsepythondict","accounts_in_message":[],"_revision_number":2},{"id":"e4576c1e2b96606090564fe2568ecf878b07cfba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 08:25:24.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/931722 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"904060c428676106d83ea041e5bc0c625aadc090","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-09 01:46:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c3bda56a30a1e5080b89d99c3e064027af2a035f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-09 01:48:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"173df4bdb61b1b15e6d05abaf53300ee871446d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 03:10:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cf3921086844841b9bb1902ced30bf0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcf3a68c1e4e4428988085f81edf5b18 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea5b2d7a6c0e459d963c2cb6482e06e3 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/94920f2dc03d4719813d001d93961c8b : SUCCESS in 4m 35s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/73ab04e2d1b74991a8c2945b25c8d0af : FAILURE in 9m 40s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dcf05af098e4807a80bc22ce0aec19b : FAILURE in 6m 38s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/939912c60d0748aaa734d03d4979329b : FAILURE in 5m 28s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29c7334fe3f3450fa26db0b4a1bd1871 : FAILURE in 6m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4f3bb2341064eccb2fa38880764d4f7 : SUCCESS in 40m 27s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6312493cdb94fb4a48f5f4515433ebe : SUCCESS in 1h 09m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23a386ce461e4b9a88b44648059cb8be : SUCCESS in 41m 47s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5480f2e492674baeb9271ad096682962 : SUCCESS in 36m 55s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/385ae8a0f9024874ac83cd7c9b081c74 : SUCCESS in 36m 43s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/db19c1d40a554e0fa9c9c89e596a0077 : SUCCESS in 35m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae87e5c609884c7ab977357d03062fb8 : SUCCESS in 1h 21m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/208d12dd93c34a61849badbcce8eda84 : SUCCESS in 58m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f12b77c3e521414784c82bd5a15307eb : SUCCESS in 57m 46s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c29214a0709409aabcb2e74412e7eb2 : SUCCESS in 57m 07s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/82893ce7fe484d03900d0d5e59414d1f : SUCCESS in 18m 11s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cf337faa5b04126953c1da1a0ec37f5 : SUCCESS in 11m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/800774784c264696ac2826d619c4a34c : SUCCESS in 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6e8b0c45877509e058bdc5102cb3b21c487de88a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-09 05:12:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a01726670e5592c8e61947500644c772645603cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 06:23:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea16d6f2df4d45a782d124ca1654d428\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55d44638f87b43eba95fb4b56848b6ce : SUCCESS in 2m 14s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ded5771f1ab74805bb1e89a74ed1122f : SUCCESS in 2m 19s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/003f9b77a7b34971a2f042c6bcf5d87d : SUCCESS in 3m 36s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/81a8cf3de2cd4e85b0e7f847b05b478e : SUCCESS in 7m 14s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f04eb15a26774076b419538f84826232 : SUCCESS in 6m 36s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d77941c49c9b4334a7d3b43742823458 : SUCCESS in 5m 31s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19d90e855ff34f8bbca2495919dca11e : SUCCESS in 6m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/072b3710ab1a494a93d3b8bab8e1ca3e : SUCCESS in 28m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f86397279c024f06961d5bb19c10c658 : SUCCESS in 1h 00m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39e88da7301f4229a2f85d9f4264452a : SUCCESS in 41m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7c091a4472a4d278705687522bb3471 : SUCCESS in 43m 53s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fb09ae0ab814aaa992a4019edb95994 : SUCCESS in 42m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a508d55b678d43c5a150471f7268e900 : SUCCESS in 55m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dacd5e18f5843aabd8136e02ed92ee6 : SUCCESS in 1h 10m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5023e49e6054c7298a56f67c0fff3f5 : SUCCESS in 49m 37s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c808f8ed565b41fdb70e47c91f8bf187 : SUCCESS in 56m 04s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/27125acf579b4981867499281babccc6 : SUCCESS in 41m 35s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ea9fd5c14f14245a90e89bccb9c3693 : SUCCESS in 16m 35s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1b9140f388d4705ac00af7a28254e7e : SUCCESS in 9m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0581ad890fa848f9b0e9b176075d868c : SUCCESS in 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f411f97e3ca2135d2d51e53ada20a865e36b27d8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-09 07:36:18.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"664e436dab6592c6de32c7127aca4634f4e2cfda","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-10-09 14:15:54.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b3c00426aaea0cca19c759b0a44183d15fd5ce9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 14:16:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"71ac63aa6fee13daba0074da820a8a61de514596","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 15:27:56.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29b5902083d448ada29056c96b8fa32d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19d445e9e6074f2c8b6a47c2f024f77a : SUCCESS in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bba59c61cf7845a5a999cd85fe82bc4b : SUCCESS in 2m 45s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/75daee57e0a84086acf9eabf6a400ee1 : SUCCESS in 9m 08s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/680647c8166a46dbba1ace43c61dd742 : SUCCESS in 6m 31s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfe8898e46eb4e8f92307f9e0a06c67a : SUCCESS in 55m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/146968fcdb484272bb58466763f7a3ba : SUCCESS in 1h 10m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e294cbec0bbd41a08ba3105124212bdb : SUCCESS in 57m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/290724165f1c4d678d70f1164d4b4e9d : SUCCESS in 58m 36s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb129614face47dfa14339fd158b3b5e : SUCCESS in 57m 41s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f484e0d7ab5540bdbca175214eba12b0 : SUCCESS in 17m 30s","accounts_in_message":[],"_revision_number":5},{"id":"20e3b94deb15b2409a2e2cae876240b6907dd3be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 15:27:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"579d10a062b77f044855fd0a524b11e6d2ecc3aa","revisions":{"3fa464491cbec33f707e3332c816a9e40ea19063":{"kind":"REWORK","_number":1,"created":"2024-10-08 08:08:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/931744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/44/931744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/1"}}},"commit":{"parents":[{"commit":"07d91b2544d9fc7d1aca23b36c1c49d9c287a614","subject":"Merge \"Drop compatibility with openstackclient \u003c 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/07d91b2544d9fc7d1aca23b36c1c49d9c287a614"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 08:08:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 08:08:13.000000000","tz":540},"subject":"Use common function to parse python dict","message":"Use common function to parse python dict\n\nDepends-on: https://review.opendev.org/931722\nChange-Id: I69dcb3bb3b479248ff5e2e84ec93a4b29b9dba22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3fa464491cbec33f707e3332c816a9e40ea19063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/3fa464491cbec33f707e3332c816a9e40ea19063"}]},"branch":"refs/heads/master"},"56bea2442a1b07e095a6f7ccd82e183d4bc0756c":{"kind":"REWORK","_number":2,"created":"2024-10-08 08:18:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/931744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/44/931744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/2"}}},"commit":{"parents":[{"commit":"07d91b2544d9fc7d1aca23b36c1c49d9c287a614","subject":"Merge \"Drop compatibility with openstackclient \u003c 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/07d91b2544d9fc7d1aca23b36c1c49d9c287a614"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 08:08:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 08:18:20.000000000","tz":540},"subject":"Use common function to parse python dict","message":"Use common function to parse python dict\n\nDepends-on: https://review.opendev.org/931722\nChange-Id: I69dcb3bb3b479248ff5e2e84ec93a4b29b9dba22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/56bea2442a1b07e095a6f7ccd82e183d4bc0756c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/56bea2442a1b07e095a6f7ccd82e183d4bc0756c"}]},"branch":"refs/heads/master"},"8514640d84ff85e1991a78f20b18f671f63784b0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-09 01:46:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/931744/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/44/931744/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/3"}}},"commit":{"parents":[{"commit":"576707bee707bb701dd20c7a3be700a030ee3286","subject":"Merge \"Use only Hash value for properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/576707bee707bb701dd20c7a3be700a030ee3286"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 08:08:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 01:46:48.000000000","tz":0},"subject":"Use common function to parse python dict","message":"Use common function to parse python dict\n\nDepends-on: https://review.opendev.org/931722\nChange-Id: I69dcb3bb3b479248ff5e2e84ec93a4b29b9dba22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8514640d84ff85e1991a78f20b18f671f63784b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/8514640d84ff85e1991a78f20b18f671f63784b0"}]},"branch":"refs/heads/master","description":"Rebase"},"2ba4a49f4523e30d1f666e9136acdbbfbd12c844":{"kind":"REWORK","_number":4,"created":"2024-10-09 01:48:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/931744/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/44/931744/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/4"}}},"commit":{"parents":[{"commit":"576707bee707bb701dd20c7a3be700a030ee3286","subject":"Merge \"Use only Hash value for properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/576707bee707bb701dd20c7a3be700a030ee3286"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 08:08:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 01:48:04.000000000","tz":540},"subject":"Use common function to parse python dict","message":"Use common function to parse python dict\n\nDepends-on: https://review.opendev.org/931722\nChange-Id: I69dcb3bb3b479248ff5e2e84ec93a4b29b9dba22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2ba4a49f4523e30d1f666e9136acdbbfbd12c844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2ba4a49f4523e30d1f666e9136acdbbfbd12c844"}]},"branch":"refs/heads/master"},"579d10a062b77f044855fd0a524b11e6d2ecc3aa":{"kind":"REWORK","_number":5,"created":"2024-10-09 05:12:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/931744/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/44/931744/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/44/931744/5"}}},"commit":{"parents":[{"commit":"576707bee707bb701dd20c7a3be700a030ee3286","subject":"Merge \"Use only Hash value for properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/576707bee707bb701dd20c7a3be700a030ee3286"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 08:08:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 05:12:07.000000000","tz":540},"subject":"Use common function to parse python dict","message":"Use common function to parse python dict\n\nDepends-on: https://review.opendev.org/931722\nChange-Id: I69dcb3bb3b479248ff5e2e84ec93a4b29b9dba22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/579d10a062b77f044855fd0a524b11e6d2ecc3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/579d10a062b77f044855fd0a524b11e6d2ecc3aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
