)]}'
{"id":"openstack%2Fpuppet-glance~937512","triplet_id":"openstack%2Fpuppet-glance~master~If6c2afd67f3a64ede35ceaab3cac9124c491a387","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-13 16:10:57.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"If6c2afd67f3a64ede35ceaab3cac9124c491a387","subject":"Deprecate support for glance-cache-(cleaner|pruner)","status":"MERGED","created":"2024-12-11 07:15:04.000000000","updated":"2026-03-13 19:16:43.000000000","submitted":"2026-03-13 19:15:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937512","meta_rev_id":"848e4baf076084230f9eae827acee57c211884a0","_number":937512,"virtual_id_number":937512,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-13 19:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-12-11 07:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-12-12 18:13:47.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-12-12 18:13:47.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-11 07:25:52.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-12-11 08:34:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-12 18:13:47.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2024-12-13 17:02:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"CC"}],"messages":[{"id":"4f2bfb38ebcdbbe6278644d1630caee2dc67de0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-11 07:15:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb681129942645ea7a0c08e9365cd63009b116a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-11 07:25:10.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8ac84e035227bc8dee8b7ece283c83de8625cf67","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-12-11 07:25:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a37f1f78fc4eea6039cb9c8031daa2a084697789","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 08:34:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ecc176967b043569737e3b567086fac\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14fbb7bbdf624d14afbd555e9cb6415a : SUCCESS in 2m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79d5171befab4254b0e1aa37a7f0c5f5 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/55ddbd3a4a474279a2a2612173b946ba : SUCCESS in 6m 44s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c6aac2832bc44308227aa7b2fc69821 : SUCCESS in 9m 43s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d10849e06b5f4e40bd22bd96b797b68f : SUCCESS in 6m 31s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/537778b379904ee7a31cfacc22431b2b : SUCCESS in 9m 39s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed6aab4788df4f8b8c9616085640278c : SUCCESS in 6m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4db8872c736740ab9bcaf14ab5c6d761 : SUCCESS in 46m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0c0d55f29f14c4cad0b212d171bd967 : SUCCESS in 1h 01m 39s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0eec7c43204a41b2b79ab2ecf48a253b : FAILURE in 31m 40s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05f4473b5473477d9673bb13173eebe5 : SUCCESS in 39m 17s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d87da55e0f24711953d7a076802c88e : SUCCESS in 39m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abd44843dced4444970c48d60726d712 : SUCCESS in 1h 07m 59s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/08effe597d8b483bb154d9ce9441ac67 : SUCCESS in 42m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/20c5ccd53e9f4916b7fb18d8a2202d37 : SUCCESS in 32m 47s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c480d404f7774e349ba2de79f0755a56 : SUCCESS in 30m 04s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/31fca07a411442109af4045d7237b1a0 : SUCCESS in 35m 02s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/1296fb507ead4134aec42587b52baaa0 : SUCCESS in 10m 46s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efc99629a2354b468e2c34490a993606 : SUCCESS in 7m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2e909aa76744613bb1e8fd3836a0d86 : SUCCESS in 6m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fce8045fe4541e392614e73a3e8bdc5 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":2},{"id":"9073077f14bfa684764b21e3ce799d0ec36cbd59","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2024-12-12 18:13:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a0f6c6ca0cf7719af609db085580de3b0b17be1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 18:14:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4a5f361906d4e799e7c770a034df81f\n\n\nWarning:\n  Change 933755 in project openstack/glance does not share a change queue with 937512 in project openstack/puppet-glance","accounts_in_message":[],"_revision_number":2},{"id":"926cf627c5c628dc1bec2c38aa3ff2af152e22cc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-12-13 17:02:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d17b5fa4f191ee54f315fc21d94ae8c25ca3a5d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 18:40:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c4423a97b0241fa958f5d4f65660d4d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82469d58e18548c3bd0625eac898364d : SUCCESS in 1m 52s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa34a00144a14fb3b7c93888336dd1df : SUCCESS in 1m 51s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6118c6bf3d84a9d91331524abd34c3d : SUCCESS in 3m 20s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/77980175a3584138835ac86b8ee227e4 : SUCCESS in 9m 12s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b06e3a9d5fe4ccbb15635a30653b4c7 : SUCCESS in 6m 15s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e70b88369c59434bb9be933848a5c398 : SUCCESS in 9m 16s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/076cf598363f4a458178effdf9a8196e : SUCCESS in 6m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3426c3c631a24f4783e47cb84f353cc2 : SUCCESS in 42m 24s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c51e05ae84ed402087750f84a59c742f : SUCCESS in 52m 09s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d301b1606770481e94778c7394c456bd : SUCCESS in 39m 44s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bd08410d40c467d99604e0020cec1a3 : SUCCESS in 37m 22s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d080f39d5f9453c86a8fcf61f4f211a : SUCCESS in 37m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a1beb0ede684beb851be2a2b238a611 : SUCCESS in 56m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/73ec7bbcbb2b4042a5a9648c9ec514e7 : SUCCESS in 1h 26m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e4cd75c764a402aa0222d1fd6bd3e6a : SUCCESS in 34m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/70bdb74cc4ad4e51a8043aae280dd2d3 : SUCCESS in 51m 37s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4330f51de169426aad401a1b715bd561 : SUCCESS in 56m 03s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/dff75401499a42da9a8a6f70f3bb85eb : SUCCESS in 9m 19s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca54bebf4a654e699e606bde09c511e6 : SUCCESS in 11m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18d95dc2ec0a4f979594d988143c7052 : SUCCESS in 10m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17aaeacd16384c88bf5fefe57e93fc03 : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4dd2eb7fe7bb78219ac9aee103794fab2165e53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 18:40:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92888a1a4946457ead40647f1b568fe3\n\n\nWarning:\n  Change 933755 in project openstack/glance does not share a change queue with 937512 in project openstack/puppet-glance","accounts_in_message":[],"_revision_number":2},{"id":"7674d1d852c8d743c7bc2d25163133c47cf38609","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-01-05 21:01:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c6126ee7db6213676f71388a1a0a649e52691d0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-01-05 21:01:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da0bf98fc15e6e55a76b75b04f4296d96ed55f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-05 22:14:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72617362f9ac45639638c73062f292f1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33108e74bce44a33bc0c2d9a0885d27c : SUCCESS in 2m 22s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f85ec4f184ed49088069706c7f0dd5dc : SUCCESS in 2m 02s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a15bab2fd8d6474aa8d7f055473d20dd : SUCCESS in 4m 44s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b890126c4ac4dde996a6f8efcab4470 : SUCCESS in 9m 29s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93753b576a5e4b07bdc3b4c86d72fb7b : SUCCESS in 5m 40s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fb6afbab8a949d588d7c7c990811edb : SUCCESS in 5m 45s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/501acc3fff5d4263b741b95e59bdbf88 : SUCCESS in 6m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59cc97e91cf6498390400415ffa2bb8a : SUCCESS in 45m 33s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e19aaa8d62c84b27adb9d3a5666c60c9 : SUCCESS in 53m 46s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc27e35ca75746a9a7c0cabd9a6e9f98 : SUCCESS in 34m 57s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a24d72405504b92ab1571e2e3f6ec1a : SUCCESS in 39m 46s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a1d8b3142324b45ac313c9b7b15d6dc : FAILURE in 21m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49eb5fdd380f407398e4e3208dbf0499 : SUCCESS in 59m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e94159c239a4741b70dc1515516e453 : SUCCESS in 58m 16s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/88bd70bc5e354f9780a403636d092d39 : SUCCESS in 58m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/146143a4670040d192fcdde32a1f469b : SUCCESS in 44m 00s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/104617115ddf43dea898c9c4b488e79d : SUCCESS in 57m 56s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fb6c40808a64ebfa60e421b35a1701a : SUCCESS in 16m 59s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/198c8970ae7a455a81fb333e2bd754c9 : SUCCESS in 8m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b9e1f8e212c4f90982c7831ee37b1c7 : SUCCESS in 10m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5588d81fd13c4dafa8709aa4aedd6593 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":2},{"id":"b85f2f5e712a26313232b9edd69ab8a5c30ddc20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-05 22:14:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23f068cba508460a8f03672bd316f616\n\n\nWarning:\n  Change 933755 in project openstack/glance does not share a change queue with 937512 in project openstack/puppet-glance","accounts_in_message":[],"_revision_number":2},{"id":"1dadd007b7ab6a75ba45bbd24acd0df04c795e9a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-13 16:10:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c67c779e81c6677367cdadc59ccd3caecb18526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 17:44:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01295e5c662a44028db6a57bf9f9c7d6\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e431cbb0031245d3afdaf2cbd5f040e8 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/210ecc7a2ee746c3a1c8004f519a96d7 : SUCCESS in 3m 21s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/82593dec559146f28c8564c060a52f5d : SUCCESS in 5m 12s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ac907d49910454193d9f2bc899f040c : SUCCESS in 7m 21s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a7760f382504e1694a066a2f02b941d : SUCCESS in 8m 49s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b94250cace584f0e9acdda833ca25641 : SUCCESS in 53m 49s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fd99b25ded249309318b3160e6c44d1 : FAILURE in 1h 21m 37s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15a72f5322cb47e69000de8c06da5bb8 : FAILURE in 1h 13m 10s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ed80cafcf2249c78c5395d6d18f03f4 : FAILURE in 1h 09m 51s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6828986d65074e3aa8a92b5f41e5ca43 : FAILURE in 1h 11m 49s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bb83db1415548a39cf689de73e07e17 : SUCCESS in 1h 09m 06s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/31ccc06f3c5f4fbd8a6685564ffe6942 : SUCCESS in 1h 26m 23s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51d6d330192841faac4553fba58ad70a : SUCCESS in 28m 40s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad458aae02924639a9165c1050b7da04 : SUCCESS in 58m 11s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2c0b13b26d347728a763b2c37e8e618 : SUCCESS in 57m 08s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc7580c028d94cf1b4fbb865a4bd2384 : SUCCESS in 45m 44s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a711f45722249feb61a5d8aa15e13ec : SUCCESS in 1h 10m 06s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2915a561e0c34863accc6a030a130382 : SUCCESS in 58m 56s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54b9b38e36b04858947eca9fb3b22f85 : SUCCESS in 38m 54s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/356cfe05930a466d89f5e822b90e9b5b : SUCCESS in 40m 35s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4a7180757054728b89cd48b4d5c8315 : SUCCESS in 20m 04s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6ed4c477a4a4aa7890ba450d7c30df9 : SUCCESS in 12m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1922d65aff884b04a24afb5469840132 : SUCCESS in 10m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75f2ee5d7b7b48788810ab4050e7d590 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":2},{"id":"1a9bb4a652189da315958a9c796f5ad9ea849921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 17:45:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cf5a0b5082b16a9edc664a45ad179b36eb8106c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 19:15:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d798812475142b794e6680562871cdc\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d84536927cc4c5c9eb78509d4468e27 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d0e478eeb404172b6fba678cdae5327 : SUCCESS in 4m 05s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/57e6c1a3b2be4478b3e838ab467a1934 : SUCCESS in 5m 20s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40b129acc98e46f6aee701fa97c32d8a : SUCCESS in 6m 59s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cb9fd27c9e74c09b4f31eaf84a83da5 : SUCCESS in 1h 07m 48s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcdd62d0e6a84da088fb6a1a34c8d744 : SUCCESS in 1h 24m 32s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0beaf519a124020b56072cdbf439618 : SUCCESS in 42m 56s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d364ad683a6d46f999ea2442110f18c3 : SUCCESS in 41m 17s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ea0ad95313e465e9b50ea1a86fc042f : SUCCESS in 30m 17s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/89e7404c6e06414094bb495e82ec95d0 : SUCCESS in 20m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fbf1dbe3508433f81d0e81034621fea : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5724f2f7b88ffd23d55e0e8e7d371f5304a4bb33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 19:15:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"848e4baf076084230f9eae827acee57c211884a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 19:16:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b968b1a509e4ffa95a49fbfc020c2a3\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b43d14a95fa4b2690b19fd29be65b2c : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c79cda4f822f13dcb69d5368c1ec808c1e5a7de0","revisions":{"fd15b41cb316316a36020b66d8d2dbbf2f6de421":{"kind":"REWORK","_number":1,"created":"2024-12-11 07:15:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/937512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/12/937512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/937512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/937512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/937512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/12/937512/1"}}},"commit":{"parents":[{"commit":"7485fcb401ad59c4523b4f5d79dd1f2f0e49dfd6","subject":"limit: Add default service_name/type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7485fcb401ad59c4523b4f5d79dd1f2f0e49dfd6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-11 07:09:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-11 07:09:08.000000000","tz":540},"subject":"Deprecate support for glance-cache-(cleaner|pruner)","message":"Deprecate support for glance-cache-(cleaner|pruner)\n\n... because these CLI commands were deprecated.\n\nChange-Id: If6c2afd67f3a64ede35ceaab3cac9124c491a387\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fd15b41cb316316a36020b66d8d2dbbf2f6de421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fd15b41cb316316a36020b66d8d2dbbf2f6de421"}]},"branch":"refs/heads/master"},"c79cda4f822f13dcb69d5368c1ec808c1e5a7de0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-12-11 07:25:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/937512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/12/937512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/937512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/937512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/12/937512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/12/937512/2"}}},"commit":{"parents":[{"commit":"7485fcb401ad59c4523b4f5d79dd1f2f0e49dfd6","subject":"limit: Add default service_name/type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7485fcb401ad59c4523b4f5d79dd1f2f0e49dfd6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-11 07:09:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-11 07:25:10.000000000","tz":0},"subject":"Deprecate support for glance-cache-(cleaner|pruner)","message":"Deprecate support for glance-cache-(cleaner|pruner)\n\n... because these CLI commands were deprecated.\n\nDepends-on: https://review.opendev.org/933755\nChange-Id: If6c2afd67f3a64ede35ceaab3cac9124c491a387\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c79cda4f822f13dcb69d5368c1ec808c1e5a7de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c79cda4f822f13dcb69d5368c1ec808c1e5a7de0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
