)]}'
{"id":"openstack%2Fpuppet-glance~938371","triplet_id":"openstack%2Fpuppet-glance~master~I1317560f1e3fcf999ef65f77f3a6e5d10ccc63fd","project":"openstack/puppet-glance","branch":"master","topic":"dev_access_log_env_var","attention_set":{},"removed_from_attention_set":{"7888":{"account":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"last_update":"2025-01-05 22:21:55.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-05 22:21:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1317560f1e3fcf999ef65f77f3a6e5d10ccc63fd","subject":"add access_log_env_var parameter to glance::wsgi::apache","status":"MERGED","created":"2025-01-02 20:53:22.000000000","updated":"2025-01-05 22:22:48.000000000","submitted":"2025-01-05 22:21:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938371-dev_access_log_env_var","meta_rev_id":"82ec5f101f9b96e1e68a75f7b9e2075849ed3b9b","_number":938371,"virtual_id_number":938371,"owner":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-05 22:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-01-03 21:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-01-05 20:56:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-01-05 20:56:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-02 21:57:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-03 01:18:19.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-01-03 21:38:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-01-05 20:56:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"9fd3599e2b6ebd307de0f084998bd88a71bfad80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2025-01-02 20:53:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3298855b0efccba63e100610876eec772d2417dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 21:57:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5712e469a2e4c4db2073c9564952007\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67f404c1ba984d93af2c47190d587313 : SUCCESS in 1m 53s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59dbf614d50b4d579ce95b665271917f : SUCCESS in 1m 54s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9213ff8d148468a98bfe59dc6383fdb : SUCCESS in 5m 39s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/70f2e75b16a34e8a89525546f4c6fbf8 : SUCCESS in 6m 53s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eea562c82f5b47519b1df9be58fb505d : SUCCESS in 6m 37s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/02925d6c555f4501bb28c7e5adff58ac : SUCCESS in 6m 31s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5ed343c4cce4b4b9c1ef8c4b3ad8324 : SUCCESS in 6m 18s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a226bb170cda4aa883ff6f2d18d6cb9b : SUCCESS in 48m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b841d274f634bb5bcf94ca686e90d31 : SUCCESS in 41m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf05d4f7ed0c43b7974fdadb05bbfe68 : SUCCESS in 37m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55bb70e2b18e48d89854e2e59a685aca : SUCCESS in 31m 16s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1aebd2104d2c45b5b764223426c99086 : SUCCESS in 34m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c677e1db91b44919f6a435ca3378ceb : SUCCESS in 51m 24s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5012821307244b38790f62a09eb077c : SUCCESS in 1h 01m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae3ec0e0a59e43f0b8f55929f8533eac : FAILURE in 50m 39s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a182b56b2804a1e869e96350cf89839 : SUCCESS in 45m 17s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/993a0ccdf75c4c52b9288c570c11c59f : SUCCESS in 47m 49s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f079be27b174ef09a9b9e7bdf2dd42f : SUCCESS in 18m 28s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ce2884f9fdd4b5a9f51e299f661a4c3 : SUCCESS in 10m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/685b858ab70b40da9e30a8f7c7650c5f : SUCCESS in 9m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05d506f51319e0f515d3dca4feb2e12da1ef9708","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-03 01:18:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88d6cc3c68cc17b2c1c5e8a0e5744e8f32aae95c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2025-01-03 08:37:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"487e8583a44ffb058d2de322a648ddf93d51e527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 09:53:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb474720491e4fdd8476b4d3fe545867\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbcdc1ac32154d16ad1baa2332d69258 : SUCCESS in 1m 56s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c370fff398b4317a90cb91a50d31aaa : SUCCESS in 2m 07s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6b050adebf842dea9a0f2c8af325715 : SUCCESS in 5m 24s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/326e8190b1fc492d9b1a4d2bcc16e0f5 : SUCCESS in 9m 43s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6b9acb0d6234909a4baa2c172fdaf2f : SUCCESS in 5m 53s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3ba7c1ba1294ff582909d5556e03e33 : SUCCESS in 6m 02s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbb07cb96b3142b8948d8ba5173e5289 : SUCCESS in 6m 31s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3052c7fa9cd43c4aa1ff86a952fbfb9 : SUCCESS in 47m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0321f82c971a451fa765a1c874d47d38 : SUCCESS in 57m 17s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bdfcbf9c8984867b6490ba2f479e891 : SUCCESS in 25m 07s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dd3bde655e840fdb13381208ba02f7e : SUCCESS in 39m 25s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e224e01bbd14663b2d1b544d33f8043 : FAILURE in 27m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/379aa579309040d483a7b0b9462ea537 : SUCCESS in 1h 10m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e14a7f879f649618d6fae9faea0e86d : SUCCESS in 47m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f46aea41beaa4003af25d952d218c741 : SUCCESS in 1h 02m 28s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca15b62fbf2f4a239af29544cca6ef25 : SUCCESS in 55m 37s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb2b696590df48e0bfd239e61ae1999a : SUCCESS in 53m 45s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5f7978aafea474080a558f4d5aec932 : SUCCESS in 19m 47s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f3576008ecf4c85ae08ef18e0e18a47 : SUCCESS in 13m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5261257bc5604f8b9ac947f5692c4dcd : SUCCESS in 11m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/394491b9d4864ce1a46097e645929768 : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2c49c30c366f154a723518923241697c1412e676","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-01-03 21:38:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"135a812aea3e0538e0eeb60d368a6a003109dc5c","author":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"date":"2025-01-04 09:48:56.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d3216970a23f4945825a5f4c29ad8e0b8318b102","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-01-05 20:56:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"816b87e5b4ce98a9882fbdcf1f53bd9378ed7fd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-05 20:57:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"949dec756c05e19b8418f3547aa5290384bbb038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-05 22:21:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4fdbcc6aa5b4c4c85a27862dc1a299f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c85a5dd36f54fdcb0c432f96a84794e : SUCCESS in 2m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6d48ffd1f0d48de891ae42e1526d50e : SUCCESS in 3m 09s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/a674a9180b0a440281472d9abe94eed9 : SUCCESS in 9m 35s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5b718f41c454230ac42461cdcad6713 : SUCCESS in 4m 08s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7470856248e34cd7be3a81d32eca2500 : SUCCESS in 48m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ced3d2b2e004410999235cf7def222ab : SUCCESS in 1h 15m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a16e326dad3b40439abe872c9d1eafa4 : SUCCESS in 48m 24s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/78e03d925750450283f2bee8bd2b44d6 : SUCCESS in 45m 06s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/db40a73508884862837da28f2cda2c3b : SUCCESS in 54m 45s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/761102f4317648ddaa14e771498ea416 : SUCCESS in 13m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9d6129e21404a658fa644c2eb09950c : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9e815288e3ffc02fbcd742bb2002a9c0549e3bd8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-05 22:21:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"82ec5f101f9b96e1e68a75f7b9e2075849ed3b9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-05 22:22:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b25fd932cf3f49149948f5f354ffffd2\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19719cec63c34ebbadd0a41754804e31 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"71f0a2146c2189448844a17ad1b3053c8974a2da","revisions":{"6743cfdf7ac9ce6fbc34d78c60a2b518d3501c66":{"kind":"REWORK","_number":1,"created":"2025-01-02 20:53:22.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/71/938371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/71/938371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/938371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/938371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/938371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/71/938371/1"}}},"commit":{"parents":[{"commit":"7485fcb401ad59c4523b4f5d79dd1f2f0e49dfd6","subject":"limit: Add default service_name/type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7485fcb401ad59c4523b4f5d79dd1f2f0e49dfd6"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-01-02 20:51:42.000000000","tz":60},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-01-02 20:53:19.000000000","tz":60},"subject":"add access_log_env_var parameter to glance::wsgi::apache","message":"add access_log_env_var parameter to glance::wsgi::apache\n\nChange-Id: I1317560f1e3fcf999ef65f77f3a6e5d10ccc63fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6743cfdf7ac9ce6fbc34d78c60a2b518d3501c66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6743cfdf7ac9ce6fbc34d78c60a2b518d3501c66"}]},"branch":"refs/heads/master"},"71f0a2146c2189448844a17ad1b3053c8974a2da":{"kind":"REWORK","_number":2,"created":"2025-01-03 08:37:16.000000000","uploader":{"_account_id":7888,"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","username":"trefzer"},"ref":"refs/changes/71/938371/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/71/938371/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/938371/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/938371/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/71/938371/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/71/938371/2"}}},"commit":{"parents":[{"commit":"7485fcb401ad59c4523b4f5d79dd1f2f0e49dfd6","subject":"limit: Add default service_name/type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/7485fcb401ad59c4523b4f5d79dd1f2f0e49dfd6"}]}],"author":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-01-02 20:51:42.000000000","tz":60},"committer":{"name":"Benedikt Trefzer","email":"benedikt.trefzer@cirrax.com","date":"2025-01-03 08:36:18.000000000","tz":60},"subject":"add access_log_env_var parameter to glance::wsgi::apache","message":"add access_log_env_var parameter to glance::wsgi::apache\n\nChange-Id: I1317560f1e3fcf999ef65f77f3a6e5d10ccc63fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/71f0a2146c2189448844a17ad1b3053c8974a2da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/71f0a2146c2189448844a17ad1b3053c8974a2da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
