)]}'
{"id":"openstack%2Fpuppet-glance~942318","triplet_id":"openstack%2Fpuppet-glance~master~I778421c9e46e3c1800dd6e7b7b210442663454f0","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-20 08:14:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I778421c9e46e3c1800dd6e7b7b210442663454f0","subject":"Fix assertion of heartbeat_in_pthread parameter","status":"MERGED","created":"2025-02-20 00:49:18.000000000","updated":"2025-02-20 08:14:49.000000000","submitted":"2025-02-20 08:14:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942318","meta_rev_id":"97973f87adc22daeedf7852b1997cb3f529da4bd","_number":942318,"virtual_id_number":942318,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 08:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2025-02-20 04:58:57.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2025-02-20 07:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2025-02-20 07:52:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 01:12:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-20 04:58:57.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-02-20 07:52:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"e4f31526abc043d0403483753ada3c1bb4ca2474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-20 00:49:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccf82f3b8049cd294eb89b2fe52b5e2b00a988d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 01:12:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cd4ce22b04f49c49928a8467986fe26\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69ea1702b5e34796bbb7edac48d5dec1 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/622d3af785204310881b1dd3e5680068 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2411365bfcf34b358d694cc2d0b6739b : SUCCESS in 5m 56s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/fba572b171fe4101a12ab7e5031d272a : SUCCESS in 7m 06s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aaf376ee65e24d3c88f2ac3c84349430 : SUCCESS in 6m 05s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/7746d26184eb407fa2024f1e7da954cb : SUCCESS in 10m 02s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6656f590e9e4e01b6ad2b7c572b5533 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d47333fc5198172bc45fb843c1d68dea30857b35","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-02-20 04:58:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6afdcf32356aede6966a8a7ce004e757c49a1695","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-20 07:52:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b460853711517617bd588e84e10da568cea8f33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 07:53:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d906c13babbdbfdf19eadcf34e9812b41b673bb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 08:14:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0a42f75bba64d7ea30ce1329f33ebae\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76211e744ffa44d3ae824e070bd83f5e : SUCCESS in 1m 53s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffdbdadbefdb43098b6012e418341dc9 : SUCCESS in 2m 01s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/9aebd745a0594498a76d2e7e90318925 : SUCCESS in 8m 09s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a526c3d5554422eb2436382a154daf2 : SUCCESS in 7m 14s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa532480389a40598584a716a1cfa823 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"97973f87adc22daeedf7852b1997cb3f529da4bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 08:14:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1a95cceb45822a33090d4afed52bdc8d3e6f34a","revisions":{"a1a95cceb45822a33090d4afed52bdc8d3e6f34a":{"kind":"REWORK","_number":1,"created":"2025-02-20 00:49:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/942318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/18/942318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/942318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/942318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/18/942318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/18/942318/1"}}},"commit":{"parents":[{"commit":"cc87de76eaa30b06db263dc60f26b05637640e9b","subject":"Merge \"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cc87de76eaa30b06db263dc60f26b05637640e9b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-20 00:49:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-20 00:49:08.000000000","tz":540},"subject":"Fix assertion of heartbeat_in_pthread parameter","message":"Fix assertion of heartbeat_in_pthread parameter\n\nIts default in puppet-oslo has been changed by [1].\n\n[1] 06a9b12199ed03842bb935749dedf4fc8e20d043\n\nChange-Id: I778421c9e46e3c1800dd6e7b7b210442663454f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a1a95cceb45822a33090d4afed52bdc8d3e6f34a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a1a95cceb45822a33090d4afed52bdc8d3e6f34a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
