)]}'
{"id":"openstack%2Fpuppet-glance~943841","triplet_id":"openstack%2Fpuppet-glance~master~I9b4296791b38804b2d1a115f6b3d8e511164fbcd","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-10 08:37:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b4296791b38804b2d1a115f6b3d8e511164fbcd","subject":"Fix new lint errors","status":"MERGED","created":"2025-03-08 16:13:12.000000000","updated":"2025-03-10 08:37:24.000000000","submitted":"2025-03-10 08:37:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943841","meta_rev_id":"ed84616f5390f7a36b1b191aa23116a81d362abc","_number":943841,"virtual_id_number":943841,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-10 08:37:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2025-03-09 23:16:01.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2025-03-10 07:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2025-03-10 07:26:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-08 17:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-09 23:16:01.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-03-10 07:26:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"e73c2523cb88c4a677ea112ee6cc0d6f1204ff7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-08 16:13:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4431bd698b6a74144289a3cc8e601ef1ab72c865","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-08 16:19:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5bc8e8670daa7dc5fe44eda02ea583f53171c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 17:35:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ca074fbe8d14fc5be6d821734dec5a6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f3571e7eff9453793a72ef6d64141ab : SUCCESS in 3m 41s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c72ae43dd334418b0fe54dceb8031d6 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/da776127dd2940bf9036f2691907e0bd : SUCCESS in 2m 50s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/82f40a980f5d458f86c4fdbb69d47d76 : SUCCESS in 10m 09s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dfed810120440d38569ee2565debbdd : SUCCESS in 5m 49s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc56c9157e9c45f3bb44693710452419 : SUCCESS in 8m 34s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96e25cea7faa4ad9bdcdfe5df9b9da6b : SUCCESS in 3m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4c35187ca294328b251e266aafb4a8a : SUCCESS in 45m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ea28acc8af74f5f984afa9cf7b6f9c5 : SUCCESS in 56m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74adf63f905f446d982152db1625f6b7 : SUCCESS in 43m 34s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/503ba0e0dabf4bc883743411e6e1a160 : SUCCESS in 41m 54s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9b7f1f41b6247fb92f9d844b34516c6 : SUCCESS in 38m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8df212ec337417a886ad35e3f5c3aaf : SUCCESS in 57m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4c936f52aa74bde960c15d160494d73 : SUCCESS in 1h 15m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/205e4d0afec0465a8d8d729c84edbe9c : SUCCESS in 25m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bee266a2537e49be94af70b392c350db : SUCCESS in 32m 43s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/95c416794ee54d0ba0610bdf82158ba8 : SUCCESS in 51m 06s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8fa186e78ea24273a34aec007b95a26c : SUCCESS in 1h 03m 50s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/74fe1cacc35f4e52960eb5ae7620c834 : SUCCESS in 1h 13m 43s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c8f5c9328424d1fa9a53847b5949dc8 : SUCCESS in 59m 12s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d238c52078e4e3e8c6614ea936c3a67 : SUCCESS in 1h 04m 26s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/02e9e2384e2c4fa79686e5f78b4bd301 : SUCCESS in 51m 19s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3258e263d4564cb69ab43754d7eeb828 : SUCCESS in 17m 26s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09ce7fad77b84ad9b8cef153e3c5da5a : SUCCESS in 11m 22s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfef307a412b4a38bace917b2bb54760 : SUCCESS in 15m 04s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e83dc24918b542dc9d307d60483be5be : SUCCESS in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72655ce082d5d778919944daf54e4c6f311db970","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-03-09 23:16:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bed8473dd4b762c340e5f935e13e78608e71703d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-10 07:26:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6f477a8c2cd99e117a25b8194c66e17c28b4d664","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 07:27:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d8eeba6189c02eb7ecb279e98a91739e466c528f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 08:37:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd282f6a978d4b8f80316aae4a25b068\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fc2cfa1c22f4c028e064786d8a272e6 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a47c8be50dc45b8b4cef11d10e78338 : SUCCESS in 5m 07s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/75c0fa4319534f63b577fff4eb340272 : SUCCESS in 4m 37s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5c327abe5034c5f9cff855beac6e854 : SUCCESS in 5m 11s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0157a3b4bfd24de4b60dc682df0af225 : SUCCESS in 6m 39s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ca77199926a474aae7d6c3225812492 : SUCCESS in 1h 05m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/346cf8e0379f4cd385e907be4f409924 : SUCCESS in 42m 24s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8aa6034521a4d3c8e8d14b59974b286 : SUCCESS in 30m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a2302e42420455dbf0b64061251d355 : SUCCESS in 30m 20s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/751964dd17b94a2eb5300d33999d5006 : SUCCESS in 34m 32s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ee3d0489d944bb08821fe02af942419 : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ed84616f5390f7a36b1b191aa23116a81d362abc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 08:37:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"197afbbf9941e5d48a99e462d8c9e18d5e83ae80","revisions":{"e8047657d5a46de1a3f329aba7fcdb19c4778541":{"kind":"REWORK","_number":1,"created":"2025-03-08 16:13:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/943841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/943841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/943841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/943841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/943841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/943841/1"}}},"commit":{"parents":[{"commit":"0b4cb4893a369461cba541fa51c34d7881237594","subject":"Add support for [image_format] require_image_format_match","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0b4cb4893a369461cba541fa51c34d7881237594"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-08 16:13:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-08 16:13:03.000000000","tz":540},"subject":"Fix new lint errors","message":"Fix new lint errors\n\nFix \"WARNING: there should be a single space before \u0027\u003d\u003e\u0027\" newly\ndetected by new puppet-lint.\n\nChange-Id: I9b4296791b38804b2d1a115f6b3d8e511164fbcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e8047657d5a46de1a3f329aba7fcdb19c4778541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/e8047657d5a46de1a3f329aba7fcdb19c4778541"}]},"branch":"refs/heads/master"},"197afbbf9941e5d48a99e462d8c9e18d5e83ae80":{"kind":"REWORK","_number":2,"created":"2025-03-08 16:19:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/943841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/41/943841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/943841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/943841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/41/943841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/41/943841/2"}}},"commit":{"parents":[{"commit":"0b4cb4893a369461cba541fa51c34d7881237594","subject":"Add support for [image_format] require_image_format_match","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0b4cb4893a369461cba541fa51c34d7881237594"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-08 16:13:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-08 16:19:21.000000000","tz":540},"subject":"Fix new lint errors","message":"Fix new lint errors\n\nFix \"WARNING: there should be a single space before \u0027\u003d\u003e\u0027\" newly\ndetected by new puppet-lint.\n\nChange-Id: I9b4296791b38804b2d1a115f6b3d8e511164fbcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/197afbbf9941e5d48a99e462d8c9e18d5e83ae80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/197afbbf9941e5d48a99e462d8c9e18d5e83ae80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
