)]}'
{"id":"openstack%2Fpuppet-glance~944305","triplet_id":"openstack%2Fpuppet-glance~master~I8f6844bf3d46721bee4b85a3740e76325d4dc736","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-14 16:06:17.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2025-03-14 04:12:28.000000000","reason":"\u003cGERRIT_ACCOUNT_21129\u003e replied on the change","reason_account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}},"hashtags":[],"change_id":"I8f6844bf3d46721bee4b85a3740e76325d4dc736","subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","status":"MERGED","created":"2025-03-14 00:47:34.000000000","updated":"2025-03-14 16:07:22.000000000","submitted":"2025-03-14 16:06:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944305","meta_rev_id":"d3bb79bea4889756b58057257a53250ef207b717","_number":944305,"virtual_id_number":944305,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-14 16:06:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2025-03-14 04:12:28.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2025-03-14 14:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2025-03-14 14:43:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-14 02:18:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-14 03:40:16.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-03-14 14:43:24.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"4bd9b3ba1917c8cdf4c97ff725e61bad749420e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-14 00:47:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f8456b22835053621fe47b5e587a36acda32d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 02:18:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06289f0ec72d4a959a84329e1ba5ab19\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da3995e3f51f455cb3222880470dbc4f : SUCCESS in 2m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59ef6edc152c4f34a6194f81ca7e7887 : SUCCESS in 2m 21s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/226515bbb9b840ff98cbe1efd402d2d8 : SUCCESS in 5m 33s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/04b3f48887df4e559ca07ca4a309221e : SUCCESS in 7m 47s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/082041f447794139a72f3e5e5f2e8c21 : SUCCESS in 3m 46s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7ea2621558c46628c6410d2a0279160 : SUCCESS in 4m 40s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f53af4e87f1c4f37b0cf01f615390e36 : SUCCESS in 6m 36s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/604f01739c37401f918aa4462b8c5948 : SUCCESS in 46m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/520b5dc945684565ae7e2ea2d897f4d9 : FAILURE in 1h 14m 45s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a10ff7fafc7a4b1d8ed5b4c0e447110a : FAILURE in 31m 30s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfa928b0382c4d60a4f1b481d1f95bd0 : SUCCESS in 42m 02s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bde169e713b4ffcabf36041f1d88ad0 : SUCCESS in 41m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bb5da64aa4b41dfb43c356db0661e91 : SUCCESS in 1h 00m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/86681b34ec26492d8bb7331811f37431 : SUCCESS in 1h 21m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e132336d9bd4ee1ae600841e1e563b4 : SUCCESS in 57m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e160668230e480eb5dc251177233268 : SUCCESS in 57m 01s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/860e06760f374bc1b83f74534ca3bf6b : SUCCESS in 58m 36s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/799a9741b9ac46beba2d4abb48ef79e9 : SUCCESS in 1h 02m 44s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f3cd0d083c8469b88f651c843e71009 : SUCCESS in 1h 18m 06s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4782e51666fe44cea02742ae50d08ef1 : SUCCESS in 1h 01m 42s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/17b780f460d3478fb4766aaa9116cc20 : SUCCESS in 52m 31s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/822fe935accc47738e931b9d1c4859a4 : SUCCESS in 54m 28s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3f5c06a3c9c444aa6089787d6c1adda : SUCCESS in 16m 24s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6287cf91845542b58e80a731a19dc330 : SUCCESS in 8m 29s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a425f0034cf442bb72b1a841ceb05a0 : SUCCESS in 17m 10s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0c179fcc94c4ba8a79e57cd5d61da5f : SUCCESS in 6m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94e7dc3e6c764e05ae9666f098503c88 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1},{"id":"129676c1801578bf6498e0eeb06698c5373b48e5","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-03-14 03:40:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99b2c2e42b7a6524a962b7706ae83cbfcb633285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-14 04:05:11.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d9acd98194d666cf7925dd91c5b3dadd0b380ef0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-14 04:05:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0b384fe18807ede4519be3a92a9f8108522d24af","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-03-14 04:12:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bd72c8ea7e207c13583b9c760f8d62475bbbd7ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 05:28:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7ed7bbff01c4660a18cef7ec1fec3a3\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/374a90fd48754a21b52ac43e0d3d651d : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/741c24d98e4647d285c0d651dd81bca1 : SUCCESS in 2m 24s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0bdc598292d4629acedb759ae06413a : SUCCESS in 6m 18s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bfe4db170f64778a2905cc04e7f7ec3 : SUCCESS in 9m 54s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c23b4b1303a4f7b96bdf60b392f11de : SUCCESS in 7m 00s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/70f177dbb09c4290909db3daf8eb75c5 : SUCCESS in 10m 07s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24a9f720035c4859a5ffd850acb90654 : SUCCESS in 3m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/869f0e1b81fb424480d262f52467edf9 : SUCCESS in 52m 15s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9e7f1b036ba4bbbba31605cc2c9be0d : SUCCESS in 26m 40s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49de5835a2564e0eb20d7787b75165c9 : SUCCESS in 19m 07s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33182727d04c43b5bf3eaf64e95e14fc : SUCCESS in 39m 03s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8302d8d0038849d4b60677b0d8a02452 : FAILURE in 59m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed01d236923147c8a61fae70cf29abce : SUCCESS in 51m 53s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3d9d5411d2942bb8bf1ff03680a0bcf : SUCCESS in 1h 17m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35ac2c423b8048e2a7c41ba132494176 : SUCCESS in 46m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d8b61e618064b86a804c47607901556 : SUCCESS in 46m 03s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ceb2c5573d004328b3c2c8b40b680a73 : SUCCESS in 43m 01s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/03794e3fa44849e38ca20158267a7039 : SUCCESS in 1h 01m 22s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6c4a1635bba4a6fb4c2178815e8e3f3 : SUCCESS in 1h 03m 15s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bce4cfb815a547d1b18f561d2f4556ea : SUCCESS in 44m 47s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3c9a1efe4e34ef297d6f2bf3d005d1c : SUCCESS in 30m 30s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51cc54d2358c4500929723792ec1ff22 : SUCCESS in 55m 11s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/44670c5cb6de43ffb90c6550b1a54c5e : SUCCESS in 14m 36s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ccf5a0594de4ca1a6e6b333f98e641c : SUCCESS in 6m 42s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f4ab542738449e9a76c040c33dd1eff : SUCCESS in 17m 29s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90bd0972752544faa50ffb56df4db09e : SUCCESS in 6m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19ce5239084d4ce2a015f6a5159d34c9 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":2},{"id":"c621522992eb6832c105e60b6efaa55995051884","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-14 14:43:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d195360a4886453f631a0b53b4ca7c769a0e6215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 14:44:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c702ecc86cb557e0e3ed8ee16a9fad29a467308b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 16:06:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/381cac268b0b42a3a7d39175253f7bf6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13181b8413464c4a892ab293af1d9723 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ae43a2730d14a38ac4c6b0a3be389da : SUCCESS in 1m 37s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab6773fcd945416480527e8ece32ce18 : SUCCESS in 6m 21s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08ebd8bb4f754f83b4eac4977f575cf1 : SUCCESS in 5m 42s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b20457ed76104a9ba1368a98b8255169 : SUCCESS in 9m 33s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8aa9cfd3e6244a338808dba83d4946e9 : SUCCESS in 41m 44s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0470b29ad524857b96947a8a9d498ed : SUCCESS in 1h 17m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/343f79871ce345498c9035ec1a200e90 : SUCCESS in 58m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be3e1fdc883949be9451fad776e99c4a : SUCCESS in 43m 40s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/018746c81eda49b981a42f9983635dd0 : SUCCESS in 50m 52s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/d91167f1f55e436181830a379e2707b0 : SUCCESS in 18m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21898302b9c74c258efebde0b6637c5f : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"41030f7a945aba23d8812422e1229ef154102cb1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 16:06:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d3bb79bea4889756b58057257a53250ef207b717","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 16:07:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c4ed961f0df49c6ae7464b863a0061f\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/363a2c7df1bb47f3bbefe87a3667296b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fa1c4754781e08259acfd098f0df282b57df03fd","revisions":{"59b7a18b19f57ec9e739b6b4b75764e9ae949d80":{"kind":"REWORK","_number":1,"created":"2025-03-14 00:47:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/944305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/05/944305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/944305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/944305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/944305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/05/944305/1"}}},"commit":{"parents":[{"commit":"197afbbf9941e5d48a99e462d8c9e18d5e83ae80","subject":"Fix new lint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/197afbbf9941e5d48a99e462d8c9e18d5e83ae80"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-14 00:46:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-14 00:46:58.000000000","tz":540},"subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","message":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params\n\nThe amqp_auto_delete option needs this patch:\nhttps://review.opendev.org/c/openstack/puppet-oslo/+/944132\n\nSince transient_queues_ttl is also managed by oslo::messaging::rabbit,\nthere\u0027s no way to use glance_api_config to set it, so this patch is\nmandatory for one to set rabbit_transient_queues_ttl in\nglance-api.conf.\n\nCo-Authored-By: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I8f6844bf3d46721bee4b85a3740e76325d4dc736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/59b7a18b19f57ec9e739b6b4b75764e9ae949d80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/59b7a18b19f57ec9e739b6b4b75764e9ae949d80"}]},"branch":"refs/heads/master"},"fa1c4754781e08259acfd098f0df282b57df03fd":{"kind":"REWORK","_number":2,"created":"2025-03-14 04:05:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/944305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/05/944305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/944305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/944305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/05/944305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/05/944305/2"}}},"commit":{"parents":[{"commit":"197afbbf9941e5d48a99e462d8c9e18d5e83ae80","subject":"Fix new lint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/197afbbf9941e5d48a99e462d8c9e18d5e83ae80"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-14 00:46:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-14 04:05:11.000000000","tz":0},"subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","message":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params\n\nThe amqp_auto_delete option needs this patch:\nhttps://review.opendev.org/c/openstack/puppet-oslo/+/944132\n\nSince transient_queues_ttl is also managed by oslo::messaging::rabbit,\nthere\u0027s no way to use glance_api_config to set it, so this patch is\nmandatory for one to set rabbit_transient_queues_ttl in\nglance-api.conf.\n\nCo-Authored-By: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I8f6844bf3d46721bee4b85a3740e76325d4dc736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fa1c4754781e08259acfd098f0df282b57df03fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/fa1c4754781e08259acfd098f0df282b57df03fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
