)]}'
{"id":"openstack%2Fpuppet-glance~954690","triplet_id":"openstack%2Fpuppet-glance~master~Ie406089d340d579790357e60defb3f9628823396","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-14 15:58:39.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-07-14 15:58:39.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2025-07-14 15:58:39.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2025-07-14 14:12:47.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ie406089d340d579790357e60defb3f9628823396","subject":"Remove swift_store_config_file","status":"MERGED","created":"2025-07-11 05:41:23.000000000","updated":"2025-07-14 15:59:20.000000000","submitted":"2025-07-14 15:58:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954690","meta_rev_id":"b3b72780e8fbd5066b9350b6fed18c6fad092383","_number":954690,"virtual_id_number":954690,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-14 15:58:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-07-14 13:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-07-14 14:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-07-14 14:12:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 07:19:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-11 14:19:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-07-14 01:50:11.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-07-14 13:03:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"1613242ced6544269217f2815aefceaa81af9890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-11 05:41:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73ea6effcedd65c3be5f5506a1b69875fc2c5b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 07:19:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56a6789edc284ebc9c9a3ed3dc82fdc5\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8f4a840529d42878b7bf42d0129026b : SUCCESS in 2m 03s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dad1f79441624cf484f5eb491efd02b4 : SUCCESS in 2m 15s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a762249959794824867c5e6655efb06f : SUCCESS in 2m 06s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/81835dfe527f4de4aaf6ce760164147a : SUCCESS in 4m 28s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70c57dfeffc442bfa0a3ea7d60a5b6d8 : SUCCESS in 2m 54s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee45f4c3a3dc47c198cf21e879120dcb : SUCCESS in 4m 31s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db2fe8414e1445e7a12c0c543509c600 : SUCCESS in 57m 16s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/340aff51eb1841cda1bbaba3e67cc397 : SUCCESS in 1h 02m 12s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/891e68a896ab4c63a8b192adade10da3 : SUCCESS in 37m 57s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/237b8a245cdc45e7bd50001057e01de2 : SUCCESS in 48m 32s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02dd4a5d8ee4440bbd1df2e033b7615c : FAILURE in 29m 34s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e63c5a57f2134da480eaebabc44a5bc5 : SUCCESS in 49m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee1290427a1c42ac8bfd36c35a58000e : SUCCESS in 1h 33m 15s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/80a33b307d214538ba593892376c1a33 : SUCCESS in 1h 00m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9623f6006f24494aa7aaa08e7545b6be : SUCCESS in 57m 29s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7d2650b0b8a4d038f9d19741668a49a : SUCCESS in 57m 58s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/94b6c11c252d4eb7a28b3fe6625356e5 : SUCCESS in 20m 48s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/662beb6339ca45ae89b8b73c8093e771 : SUCCESS in 9m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1764d8b287b46cfba3629471bf98d74 : SUCCESS in 9m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25dce29155034c07a785a5ebebbb9e88 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5884a72078afd2910d95558ab8b29bc030765ab7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-07-11 14:19:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b733a020c7a2db77ccf81f9b91786f26049ad36","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-07-14 01:50:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e473f9b625a326a98116f9d752427add93022e89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-14 02:58:45.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0d8b12a53c20d9621993383e66ae6a230780cb63","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-14 02:59:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fbc1780804ff8feb1106c786a59cb90a00ddbbe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 04:09:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/620e99a431bd46a691f944e936a1f1b3\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02e4c58750334822b42206770beda0b4 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6db5ac8341345eb9ba3aaf1251d8532 : SUCCESS in 2m 24s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40d8ada2a66748a7b44773f0c628d80e : SUCCESS in 2m 30s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/098e465ae0cb4210b20bf42c963b991a : SUCCESS in 9m 49s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bac6c40c506848ecbe8dcdd870f4f8af : SUCCESS in 6m 05s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21b4457bc11941f5a05d490485c27fa1 : SUCCESS in 4m 55s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c2f77550246438fb7e073837d76f762 : SUCCESS in 50m 06s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc62de988bdd4aab8938bc551e6e6129 : SUCCESS in 1h 06m 57s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f73da0ef40048b29133c90af3344fdc : SUCCESS in 39m 37s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d22efa0587ec4e80af6a833a95aaf547 : SUCCESS in 43m 56s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2ae1d2f45ac4980ba89c3a8c9a1ae6c : FAILURE in 24m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18aa7080e4ac41e2995e9c6eae9b826b : SUCCESS in 1h 04m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab38e1713b3c485997afbcc3f95a3d90 : SUCCESS in 1h 02m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0d2110c3ca84a5189742d686728e143 : SUCCESS in 45m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9731f1959a54becbe4ab9187872e113 : SUCCESS in 53m 52s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e0904825dcc4c8785c4a1af2e7dc3c3 : SUCCESS in 58m 04s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fe1a28830b844f48a62bc70ddcbc11f : SUCCESS in 15m 32s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07a01faedce843e7a984050265ae14ff : SUCCESS in 9m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d55b79d8707c45b0851634c502198de1 : SUCCESS in 9m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce81573673fb4b9ea9ad374dc70c4e19 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":2},{"id":"bf3aeb2f954f16f85622ec4772c934e567d739e0","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-07-14 04:34:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"365771be68ce33dd569c6ecbdab6ee191e6e11cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-07-14 13:03:10.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"320b8fed7d4b77a60d4fff5714af8edeafd11235","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-07-14 13:03:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1c486a7b77959293721f1b79f55653a28f798c97","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-07-14 14:12:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"91c92d0711a449710b1bcbd997e1f88771c7766c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 14:29:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37dfdb23fa194593926986a010f26d31\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f8226aa3be84accaee201bd8afa86f4 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfefe5c8228b40579c3ec6300041cf7b : SUCCESS in 3m 12s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fdfe085e745e4b6cb6ccea3da983d325 : SUCCESS in 2m 28s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/4efa43f33ca24c7182761c06b7da3e04 : SUCCESS in 10m 44s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9ca767580154c21bc785d21348e43ca : SUCCESS in 5m 54s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b41f279bff2407faa8026f5b8d21159 : SUCCESS in 6m 00s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4dc5287e62f493c90312d29e2fa8b13 : SUCCESS in 44m 55s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7207bb57a9d7476f8bf1f05fe3f0e12a : SUCCESS in 53m 23s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/749b0f6143374325ad61f8b6463a72c3 : SUCCESS in 40m 41s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cadfead838ec4499ade5539fdf63b814 : SUCCESS in 38m 13s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae10ce89b29844d88794abdee15d88cd : SUCCESS in 39m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5482fc131ec4abbb0710ba1d9faa98c : SUCCESS in 1h 03m 35s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/331fd1c135ea481b92e56117a4f472b1 : SUCCESS in 1h 21m 53s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ccbc1d1c847f4d68b469f5d10cf1395e : SUCCESS in 51m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2df60592a8b04d5fa77649f1992a0d97 : SUCCESS in 52m 05s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1ab4a0d4c1d439eae9d0c6244ded74f : SUCCESS in 48m 39s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/572d0b1087a34a0ea1c6ad7f7216479a : SUCCESS in 17m 34s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ac39f6c491d4f52a5af4a8ff6261185 : SUCCESS in 12m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4c1963da31b41648cade970a706fb12 : SUCCESS in 11m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50b92df7edf74f2596a72fad374f0072 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":3},{"id":"39a50429639ec64ce56b4a63765cb440a8e5baa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 14:30:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5585358978d4081485b7a0ac9e9654872c0660da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 15:58:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d062f145cc64084a3c51ff9badb9efe\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbb74b87ec1647169957338f575d329e : SUCCESS in 1m 31s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/051f808e7cfe449f98833e80e4f06e57 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ad7832e633646d7924a230e3edef858 : SUCCESS in 2m 26s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/360898d941b442e3ba7af49802986c24 : SUCCESS in 9m 43s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f283292f8ea1423592718408e7a440a4 : SUCCESS in 5m 05s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b07cd54ade594ee3aba07848439f9069 : SUCCESS in 1h 08m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8aa663baa2347f18c2a068885c0aae8 : SUCCESS in 1h 23m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3efd863bb084d26879f4f2887894f9d : SUCCESS in 26m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/46ac47fa734042e9a730bd88c3c3e382 : SUCCESS in 52m 01s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e46f9509b6e64bf18b61037f79652eb6 : SUCCESS in 42m 35s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4cfa1e43b134021827724439ed516ea : SUCCESS in 17m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10ce69630df5457ba18ac19b22637b1d : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":3},{"id":"956796141754543dd2619dd84e0bc20c24712de5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 15:58:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b3b72780e8fbd5066b9350b6fed18c6fad092383","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 15:59:20.000000000","message":"Patch Set 3:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c96d1a791c22455e929bc667415deed9\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b23a2035384486cbc3abfed23b8a66e : FAILURE in 31s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c7e0af089b632cae8445a7ca3ed4dbd2d95f4635","revisions":{"95922addf7474a3db757ba1e3cd4ce9e39006c4e":{"kind":"REWORK","_number":1,"created":"2025-07-11 05:41:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/954690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/954690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/1"}}},"commit":{"parents":[{"commit":"a93781c2d324f0d81b146436c3dacc737eb7d5de","subject":"Merge \"Deprecate support for filesystem_store_datadirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a93781c2d324f0d81b146436c3dacc737eb7d5de"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-11 05:37:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-11 05:41:05.000000000","tz":540},"subject":"Remove swift_store_config_file","message":"Remove swift_store_config_file\n\nThe parameter was deprecated 5 years ago[1], to encourage users to\nuse the default file, which can be managed by the glance_swift_config\nresource type.\n\n[1] 7f1729e9aebc20d892baaf4678221bcc24135853\n\nChange-Id: Ie406089d340d579790357e60defb3f9628823396\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/95922addf7474a3db757ba1e3cd4ce9e39006c4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/95922addf7474a3db757ba1e3cd4ce9e39006c4e"}]},"branch":"refs/heads/master"},"0d2cacf2c9ac4e2cf28159fda8bb394af7a5fbd7":{"kind":"REWORK","_number":2,"created":"2025-07-14 02:58:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/954690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/954690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/2"}}},"commit":{"parents":[{"commit":"a93781c2d324f0d81b146436c3dacc737eb7d5de","subject":"Merge \"Deprecate support for filesystem_store_datadirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a93781c2d324f0d81b146436c3dacc737eb7d5de"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-11 05:37:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-14 02:58:45.000000000","tz":0},"subject":"Remove swift_store_config_file","message":"Remove swift_store_config_file\n\nThe parameter was deprecated 5 years ago[1], to encourage users to\nuse the default file, which can be managed by the glance_swift_config\nresource type.\n\n[1] 7f1729e9aebc20d892baaf4678221bcc24135853\n\nChange-Id: Ie406089d340d579790357e60defb3f9628823396\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0d2cacf2c9ac4e2cf28159fda8bb394af7a5fbd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0d2cacf2c9ac4e2cf28159fda8bb394af7a5fbd7"}]},"branch":"refs/heads/master"},"c7e0af089b632cae8445a7ca3ed4dbd2d95f4635":{"kind":"REWORK","_number":3,"created":"2025-07-14 13:03:10.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/90/954690/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/90/954690/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/90/954690/3"}}},"commit":{"parents":[{"commit":"a93781c2d324f0d81b146436c3dacc737eb7d5de","subject":"Merge \"Deprecate support for filesystem_store_datadirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a93781c2d324f0d81b146436c3dacc737eb7d5de"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-11 05:37:56.000000000","tz":540},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-07-14 13:03:10.000000000","tz":0},"subject":"Remove swift_store_config_file","message":"Remove swift_store_config_file\n\nThe parameter was deprecated 5 years ago[1], to encourage users to\nuse the default file, which can be managed by the glance_swift_config\nresource type.\n\n[1] 7f1729e9aebc20d892baaf4678221bcc24135853\n\nChange-Id: Ie406089d340d579790357e60defb3f9628823396\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c7e0af089b632cae8445a7ca3ed4dbd2d95f4635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c7e0af089b632cae8445a7ca3ed4dbd2d95f4635"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
