)]}'
{"id":"openstack%2Fpuppet-glance~957325","triplet_id":"openstack%2Fpuppet-glance~master~Iceb0be07fa808f2016c4939bf7444ed6d9f09260","project":"openstack/puppet-glance","branch":"master","topic":"stdlib-func","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-15 01:30:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iceb0be07fa808f2016c4939bf7444ed6d9f09260","subject":"Switch to namespace sdtlib functions","status":"MERGED","created":"2025-08-14 09:04:07.000000000","updated":"2025-08-15 01:30:19.000000000","submitted":"2025-08-15 01:30:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957325-stdlib-func","meta_rev_id":"532e59e6fe0b67c65d2d06a07d8a0b8bdf323729","_number":957325,"virtual_id_number":957325,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-15 01:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2025-08-14 12:26:02.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2025-08-14 19:54:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2025-08-14 19:54:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 12:12:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-14 12:26:02.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-08-14 19:54:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"0f8ad4abec3aac89a22c90a003789e0bf4284f61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-14 09:04:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"231cbe6acb8412c80825654555292a981faef39d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-14 09:13:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"776872aa14d0ee8dbaf4606ff2ddef5ad5ae2914","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-14 09:23:09.000000000","message":"Topic set to stdlib-func","accounts_in_message":[],"_revision_number":2},{"id":"b4c3441883a25e1bb5c5897b842cb9657d1fc7bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-14 09:23:38.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c651886e838ba39fa707901321791f88e6be01ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 12:12:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73ea017500f343b986d67b061c3bfd6f\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/f16864dda0aa439ebcd729633a27a177 : SUCCESS in 1m 21s\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b7e097b76934fb4aef1f725b62bdd81 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d09fd6e83b7e40ef908cb1692f000eb6 : SUCCESS in 2m 29s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3516d86999d440b2b051375e5ffdbdcd : SUCCESS in 2m 17s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/369d1ebae6074ba8b93520ff7a3c69c0 : SUCCESS in 7m 20s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f03d5c4bd11f421c8bb75f43187a70b3 : SUCCESS in 3m 04s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b737c4de74314925a57b99624b47769c : SUCCESS in 5m 09s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39cd989d9a6e433280fc90e3e5c23982 : SUCCESS in 23m 31s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bb3f2bfb30649999a2587a4afb3e6ad : SUCCESS in 34m 07s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b05aecca969c4b37a7d54b2be52e6827 : SUCCESS in 22m 47s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8472dec037747bfafee86e10f43b852 : FAILURE in 22m 21s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23bb4091be25485c820f824c1cad83dd : FAILURE in 27m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffb8525235fb457d876f9add4760d8af : SUCCESS in 31m 47s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/20725f57c15946629cc6ac56a72997bb : SUCCESS in 1h 20m 06s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fce927ec2944d8bb712022bf1917f8b : SUCCESS in 31m 39s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/510f3c34d35a48a6aafa44f8990f04db : SUCCESS in 33m 32s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e182285d439495bac4e92916a3d8087 : SUCCESS in 57m 07s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/863afabdef98495fb102c14320287424 : SUCCESS in 10m 48s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a4e2f954d824887a8ce6b16ca04a249 : SUCCESS in 10m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9c5ffaeaa2b4bf89a3e8e95f2d68b16 : SUCCESS in 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"48af67a2bab931d1b44261a192a2906d57b27e09","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-14 12:26:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"39096a587d5e4a11ca483b8a3ea58474863b4882","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-08-14 19:54:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1d712d8721a56856b9ebf3d82ea8ac980123503a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-08-14 19:54:29.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1f09462b6c68fdb75d9029fbb136d77d64add1ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 19:55:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a5069d3339739f1e489c5e20005cfd407da71148","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 01:30:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ed297412af6426d812d32da79113ff2\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/6833ca21755b4b7fb1f16c2d2040c250 : SUCCESS in 1m 49s\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b009494ccc6457baf61849283660076 : SUCCESS in 1m 23s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a229ddb47ce34ed08f82f02f8f7f8fc7 : SUCCESS in 1m 24s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da01e6966af843e38fdade3525d2a455 : SUCCESS in 1m 31s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4edeefb30c349869a59deccd1636b54 : SUCCESS in 8m 25s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43bcc9235eab4fb5b8f233927d4a96c8 : SUCCESS in 3m 23s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/808414877dc84fcd9c2fcd7273d876f9 : SUCCESS in 1h 04m 59s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2191d8bc66348f7966d8ece70d862ac : SUCCESS in 39m 50s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8a8880da4114b1382b94663494c13b9 : SUCCESS in 26m 10s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb50ab8d49054ca188577f42a3215ab6 : SUCCESS in 28m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e439e6f1715c476aa40036caaf9b05d2 : SUCCESS in 50m 14s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/cedb24f8e54e494b92c926c4289ec8e7 : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":3},{"id":"532e59e6fe0b67c65d2d06a07d8a0b8bdf323729","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 01:30:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"88d0d028d924f6b533f91706e1df8eaea2943d2a","revisions":{"853d1b09ecb4b4288b08f9856b8b04ce901fe948":{"kind":"REWORK","_number":1,"created":"2025-08-14 09:04:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/957325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/25/957325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/1"}}},"commit":{"parents":[{"commit":"6165e24030e649853f80ecb021787a050ba6fd35","subject":"Add support for cinder backend default options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6165e24030e649853f80ecb021787a050ba6fd35"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-14 09:03:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-14 09:03:59.000000000","tz":540},"subject":"witch to namespace sdtlib functions","message":"witch to namespace sdtlib functions\n\npuppetlabs-stdlib deprecated the existing functions in favor of\nthe new stdlib:: functions in 9.0.0. The version was released a few\nyears ago and now we assume we are ready to adopt to the migration.\n\nChange-Id: Iceb0be07fa808f2016c4939bf7444ed6d9f09260\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/853d1b09ecb4b4288b08f9856b8b04ce901fe948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/853d1b09ecb4b4288b08f9856b8b04ce901fe948"}]},"branch":"refs/heads/master"},"02072777eaafacce4d53227c5b12adabf1cf30d4":{"kind":"REWORK","_number":2,"created":"2025-08-14 09:13:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/957325/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/25/957325/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/2"}}},"commit":{"parents":[{"commit":"6165e24030e649853f80ecb021787a050ba6fd35","subject":"Add support for cinder backend default options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6165e24030e649853f80ecb021787a050ba6fd35"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-14 09:03:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-14 09:13:25.000000000","tz":540},"subject":"witch to namespace sdtlib functions","message":"witch to namespace sdtlib functions\n\npuppetlabs-stdlib deprecated the existing functions in favor of\nthe new stdlib:: functions in 9.0.0. The version was released a few\nyears ago and now we assume we are ready to adopt to the migration.\n\nChange-Id: Iceb0be07fa808f2016c4939bf7444ed6d9f09260\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/02072777eaafacce4d53227c5b12adabf1cf30d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/02072777eaafacce4d53227c5b12adabf1cf30d4"}]},"branch":"refs/heads/master"},"88d0d028d924f6b533f91706e1df8eaea2943d2a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-08-14 09:23:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/957325/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/25/957325/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/25/957325/3"}}},"commit":{"parents":[{"commit":"6165e24030e649853f80ecb021787a050ba6fd35","subject":"Add support for cinder backend default options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6165e24030e649853f80ecb021787a050ba6fd35"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-14 09:03:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-14 09:23:38.000000000","tz":0},"subject":"Switch to namespace sdtlib functions","message":"Switch to namespace sdtlib functions\n\npuppetlabs-stdlib deprecated the existing functions in favor of\nthe new stdlib:: functions in 9.0.0. The version was released a few\nyears ago and now we assume we are ready to adopt to the migration.\n\nChange-Id: Iceb0be07fa808f2016c4939bf7444ed6d9f09260\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/88d0d028d924f6b533f91706e1df8eaea2943d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/88d0d028d924f6b533f91706e1df8eaea2943d2a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
