)]}'
{"id":"openstack%2Fpuppet-glance~957932","triplet_id":"openstack%2Fpuppet-glance~master~I107209fdfd7136251a94887bec6499d36fdbf9d9","project":"openstack/puppet-glance","branch":"master","topic":"lint-plugins","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-23 02:35:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I107209fdfd7136251a94887bec6499d36fdbf9d9","subject":"Prepare for voxpupuli-puppet-lint-plugins","status":"MERGED","created":"2025-08-19 14:47:33.000000000","updated":"2025-08-23 05:41:28.000000000","submitted":"2025-08-23 05:41:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957932-lint-plugins","meta_rev_id":"48506d3ca1bd314d251e3853cbf147888a17d977","_number":957932,"virtual_id_number":957932,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-23 05:41:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-08-20 06:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-08-22 14:03:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-08-22 14:03:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-19 15:13:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-20 06:39:00.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-08-22 14:03:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"03c8070df1ea1543654461009da76cf6ced04531","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 14:47:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b2bdc795eb43fae502ad1f9511e1d76ee81d595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 15:06:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42dedbb9979c328866dd154359410828c8787646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:13:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/557bd5d9f9e74571bc5de909810fa2d0\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34550d3dde2f49b9825645fda11138b0 : FAILURE in 3m 48s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c5c8e8ab09f4f65b0445f9557ae4804 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":2},{"id":"4fe8efac79c7904ff5c56eabf9d34ef6ec3bab7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 15:21:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4e6f2041a8e0ead72fcc42bb0e900bff1485c14d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 15:28:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d45498132b4b40c8be7271a066df0556\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7c511b560ed41a8a24d873bef95c640 : FAILURE in 3m 52s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e0e52bdd89d4ec896384b9a7287307b : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":3},{"id":"56b8b78fe760b96204362b617ccfd4f6011821f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-19 15:30:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9a0b54c2d761586dcdc682f110b8e8ad3282aee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 19:15:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a74524a22174bb591d48bc2421e8354\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed4a6cec27ab48e9aec479250ca48ba0 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d6f6a7d6bf04ef08a20c09995da55b7 : SUCCESS in 2m 49s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0645ff6a608844108c23c1ab413b546a : SUCCESS in 9m 37s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad4c1516a04c4a6788c74a07922e103d : SUCCESS in 7m 02s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd5f57ccbff44539b459389de73dc92d : SUCCESS in 4m 33s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03d97b2079a14e0c94bcf6bb798261d3 : SUCCESS in 48m 48s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d97ab98b6404619b803158510f1f6c4 : SUCCESS in 1h 04m 14s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12fe9d08106248bc81ca7317fc7e242f : SUCCESS in 24m 54s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d21a89f63a9546748fe83fcb39bd877c : SUCCESS in 47m 48s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/682717363f094fdca8fa96a0255bdf36 : FAILURE in 31m 30s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a5c03169c4a40eea7bc5f1d7c2ed351 : SUCCESS in 1h 06m 29s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5329583df96347ebabfa2cf73b06d64e : SUCCESS in 1h 19m 12s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/09a6b8fca6894fe697958a0a2a8e8a66 : SUCCESS in 1h 01m 08s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fecdde62cf94fa2a92ef30b330d5f6c : SUCCESS in 58m 31s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c81c76779ff44eb3af9d67f8f821c181 : SUCCESS in 1h 03m 50s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/64c587590b3a4d3b97d6c066ec0cb8bb : SUCCESS in 59m 50s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7572c0e089ae439d8652de54c3cb4f96 : SUCCESS in 1h 27m 16s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a0345aad6eb43ccacf6ab79597b5103 : SUCCESS in 51m 01s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/061670e7adeb4de6a20fddfbd8c983dd : SUCCESS in 27m 40s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/063e3a1ecbd74a368737de65215e0602 : SUCCESS in 29m 45s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/269ece6e2c4c4b30a8d3c239c2a093c6 : SUCCESS in 17m 09s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dff352e72ffb4c6c9755c37bd64ba0bf : SUCCESS in 7m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/194079f4152e413794f24ec67079f4b2 : SUCCESS in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8e97df87a81808fff58ba00b5334d995868fb4cc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-20 06:39:00.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5323865161524855c908009a83ffaeb670ee15dc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-08-22 14:03:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2e86c376649ea6c3b909d0dae41faf2f9b8d882a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 14:04:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"863001272650c380767c5a35d44375084ae19ea3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 14:23:52.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c1c175db22e415f880612849b06d090\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67301d1d24cf4aa4926ee96ed428e90b : SUCCESS in 3m 02s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf6e8b1c460b492bb0dab79a395aced4 : SUCCESS in 4m 22s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/db2b9cb5f02c4204b6d27eca2e4c8756 : SUCCESS in 9m 38s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69a1022848144f96a9a83c79a1ec4fb1 : SUCCESS in 6m 39s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6a6664da2254164a2b4c61a0d6d0cb9 : FAILURE in 17m 58s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c84903c075354736816c64aeca63e3a4 : FAILURE in 13m 41s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/095b1623bd7a4c3d9817cf370a0bdde9 : FAILURE in 13m 31s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/47e37f7af63c4bb4ae1aa9155663b0cf : FAILURE in 13m 10s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2355702bcfb74994bdee570cf45543ae : FAILURE in 13m 08s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cbd5404259441deac011ca54579dd6f : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":4},{"id":"5483acd99374192c4931f35cedd49809dff6e017","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-23 02:35:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4cd77ece01b2be2b2bf0968f50d4510662f1bc4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 04:09:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a5e0f97a9af4e14bf3ba7f9806e6b00\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb5ae89f9998443aa00f69acd8222b09 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72a9407d29a645a48ccad8b18cffc797 : SUCCESS in 2m 50s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bff8b211d3a49209cdda92f269ca015 : SUCCESS in 9m 08s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e39d594d936a4f2a9d2bcaf137487782 : SUCCESS in 5m 47s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38493a3bc0004cac8d9e3ed835b8f324 : SUCCESS in 4m 48s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97a1ad52615b4d92a8a25baa761dccec : SUCCESS in 44m 44s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/400fc028095e47d3b6621bd469681ff2 : SUCCESS in 1h 27m 13s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/759b29bd5b5241bdbbb52b44e511bee1 : FAILURE in 32m 51s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95f23cd082bf4091ae1dc85c3cad55de : SUCCESS in 49m 59s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e002a3c99aad453d8b2aad0faaf1eba6 : FAILURE in 45m 46s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ddb9befcac64c2685ad0899ca2a53c4 : SUCCESS in 56m 33s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6aad42aacecb421292bda4b81074ce00 : SUCCESS in 1h 14m 02s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/554648f393c9409cbc55e701859ebd5e : SUCCESS in 57m 12s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad1188f71b454e529c6968478549ce51 : SUCCESS in 56m 19s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/352b84bfe68a4392b8bf57a7ae61c62f : SUCCESS in 58m 45s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/68d9ace9fc71439aa3a974d365d9f7c2 : SUCCESS in 1h 09m 39s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b6a4f9b1d8444ea99fd613f32f54cd9 : FAILURE in 21m 01s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3dc7e3d07f5542a89a4edfb053dfdde2 : SUCCESS in 57m 06s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c398e7edc1c946068c5616250c4e6128 : SUCCESS in 1h 17m 28s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5657104bae5451bba5b30e69d020f9f : SUCCESS in 53m 26s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/922e70af9fbc4617ad33d1812d86d342 : SUCCESS in 18m 49s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4128b050da04489991a040a7f9b10203 : SUCCESS in 10m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4073e941f8f3404a996deb26c65f84e4 : SUCCESS in 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7e6d508b2f8b4d4d54e799f5d4e707442f5d7eb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 04:09:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"88eb164c978e0a8ce7149080be4865ba8ccbc769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 05:41:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05b4cf6afe3d44139fd121d25c599dbf\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4baf691c2ec4d5e9a546ff92855b1f6 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5dd4d0c5685410bb5919a951e6aa88d : SUCCESS in 2m 09s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/951e44749f1b4dd595e9dd9529cbaea2 : SUCCESS in 10m 13s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec355c27715242109f561a146ec86665 : SUCCESS in 6m 37s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1eaf2fef24f04c27899497f04c4c727d : SUCCESS in 33m 45s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f87dd8f935549e3b94a68ea97232e0a : SUCCESS in 1h 27m 35s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/481d20c7bbeb415e8788c754c5853e99 : SUCCESS in 26m 51s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aef70c822f2a472eb5a5fe502446f29f : SUCCESS in 58m 37s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef943ec9fbf2441a8744541ff39e41ee : SUCCESS in 59m 22s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8033046e247a40249822957034afddc8 : SUCCESS in 18m 38s","accounts_in_message":[],"_revision_number":4},{"id":"48506d3ca1bd314d251e3853cbf147888a17d977","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 05:41:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d5a380bdeca856e5fe65cec306208877d9ca0807","revisions":{"923977f592348c91a990aafa071a7c7de34aee00":{"kind":"REWORK","_number":1,"created":"2025-08-19 14:47:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/957932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/32/957932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/1"}}},"commit":{"parents":[{"commit":"9624e1eac260133e701a89ff5952c7855693544a","subject":"Remove puppet 7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9624e1eac260133e701a89ff5952c7855693544a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 14:47:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 14:47:22.000000000","tz":540},"subject":"wip: Prepare for voxpupuli-puppet-lint-plugins","message":"wip: Prepare for voxpupuli-puppet-lint-plugins\n\nDepends-on: https://review.opendev.org/957884\nChange-Id: I107209fdfd7136251a94887bec6499d36fdbf9d9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/923977f592348c91a990aafa071a7c7de34aee00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/923977f592348c91a990aafa071a7c7de34aee00"}]},"branch":"refs/heads/master"},"4eb09362d45d00ac8cff07bc276ff858a6316fdf":{"kind":"REWORK","_number":2,"created":"2025-08-19 15:06:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/957932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/32/957932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/2"}}},"commit":{"parents":[{"commit":"9624e1eac260133e701a89ff5952c7855693544a","subject":"Remove puppet 7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9624e1eac260133e701a89ff5952c7855693544a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 14:47:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 15:06:23.000000000","tz":540},"subject":"wip: Prepare for voxpupuli-puppet-lint-plugins","message":"wip: Prepare for voxpupuli-puppet-lint-plugins\n\nDepends-on: https://review.opendev.org/957884\nChange-Id: I107209fdfd7136251a94887bec6499d36fdbf9d9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4eb09362d45d00ac8cff07bc276ff858a6316fdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/4eb09362d45d00ac8cff07bc276ff858a6316fdf"}]},"branch":"refs/heads/master"},"a6e987ac806f630ecacf3e96597d05b39357860f":{"kind":"REWORK","_number":3,"created":"2025-08-19 15:21:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/957932/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/32/957932/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/3"}}},"commit":{"parents":[{"commit":"9624e1eac260133e701a89ff5952c7855693544a","subject":"Remove puppet 7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9624e1eac260133e701a89ff5952c7855693544a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 14:47:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 15:21:22.000000000","tz":540},"subject":"wip: Prepare for voxpupuli-puppet-lint-plugins","message":"wip: Prepare for voxpupuli-puppet-lint-plugins\n\nDepends-on: https://review.opendev.org/957884\nChange-Id: I107209fdfd7136251a94887bec6499d36fdbf9d9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a6e987ac806f630ecacf3e96597d05b39357860f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/a6e987ac806f630ecacf3e96597d05b39357860f"}]},"branch":"refs/heads/master"},"d5a380bdeca856e5fe65cec306208877d9ca0807":{"kind":"REWORK","_number":4,"created":"2025-08-19 15:30:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/957932/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/32/957932/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/32/957932/4"}}},"commit":{"parents":[{"commit":"9624e1eac260133e701a89ff5952c7855693544a","subject":"Remove puppet 7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9624e1eac260133e701a89ff5952c7855693544a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 14:47:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-19 15:30:09.000000000","tz":540},"subject":"Prepare for voxpupuli-puppet-lint-plugins","message":"Prepare for voxpupuli-puppet-lint-plugins\n\nFix new lint errors detected when full of the voxpupili lint plugins\nare enabled.\n\nChange-Id: I107209fdfd7136251a94887bec6499d36fdbf9d9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d5a380bdeca856e5fe65cec306208877d9ca0807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d5a380bdeca856e5fe65cec306208877d9ca0807"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
