)]}'
{"id":"openstack%2Fpuppet-glance~959270","triplet_id":"openstack%2Fpuppet-glance~master~Ib7a8d4983210c084d83599b4a69623371b09ca9b","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-03 15:02:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7a8d4983210c084d83599b4a69623371b09ca9b","subject":"Remove support for [oslo_messaging_rabbit] heartbeat_in_pthread","status":"MERGED","created":"2025-09-03 02:00:37.000000000","updated":"2025-09-03 15:03:44.000000000","submitted":"2025-09-03 15:02:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959270","meta_rev_id":"4e72e38c68200d8d34229c8891e2fd7bbd2a6e8e","_number":959270,"virtual_id_number":959270,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-03 15:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-09-03 07:36:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-09-03 13:29:14.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-09-03 13:29:14.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 03:43:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-03 07:36:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-09-03 13:29:14.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"39377fbb1a13ed3bd31c630ea0c5470f278a1f4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-03 02:00:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"304acc80b75a28187b19bfff77b417c938baa3cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 03:43:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27fd61dba5b34aaf97361d0e45d00caa\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c24847dc8cf54e319ed7a3383ec2310e : FAILURE in 3m 17s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b27d907994a64afa98bfa2882014c030 : SUCCESS in 2m 58s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/f9d1c13b5ed24ad0b28567c0f47b9b46 : SUCCESS in 4m 46s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5476666ce75f45a8bef832df2824ed01 : SUCCESS in 6m 46s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6a19e8aab484d5cab79553e5888d8c8 : SUCCESS in 5m 33s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76ccb9accaa64bbbb1216d4663229c64 : SUCCESS in 58m 56s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eaea61e3bf84488aa03e5bf68b466261 : SUCCESS in 1h 05m 23s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e47e60781f9488d9c8eca6cd941c931 : SUCCESS in 47m 30s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3af961b25bb145d8a36749db7b3154ec : SUCCESS in 33m 29s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d607ae27c2348129e36d476dcce03c0 : FAILURE in 26m 53s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc813db77f504316827986335a4a6e33 : SUCCESS in 52m 47s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/42205b82ebf346e5a6ecbc221d64c2e9 : SUCCESS in 1h 21m 33s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/913491aba67f413faaf68b868b1e992c : SUCCESS in 1h 02m 44s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f8734d3faa54205a4d945aa53b1a1ce : SUCCESS in 53m 54s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/801c42ec81f9498b985a7d4eff5a8af5 : SUCCESS in 28m 23s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cd6756bac5d4a35831135698d2809ce : SUCCESS in 1h 08m 48s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e78b9c7915a4865ac63a4c7efe95c09 : SUCCESS in 1h 20m 56s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1847e595260947c3ae96a88d053732f0 : SUCCESS in 58m 22s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c6ee16ca18f4fd1939dea4a1c0c1e33 : SUCCESS in 56m 36s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c7f8c23e23d4b9794beb646eab22a69 : SUCCESS in 49m 05s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/e787677341fe428684adc9c84accee31 : SUCCESS in 13m 21s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c0958ebfece4389905ff04f4b482438 : SUCCESS in 8m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56063cad282c44c385c604e1a42b02e7 : SUCCESS in 12m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f6fa6a5d55b46e7a87cfd496e3c3f8b : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6a318eadb323946ac434702ac13ee10adcae733e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-03 07:30:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c25cd507ee40593085afb9d9466a36d8c31c8a5e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-03 07:36:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b78d8649af1433ed64d60c139639e5962a1001d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 09:00:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5be1ff338964ad6963f9535260a888c\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2da28d5efddb4e39bb4ff9f1029e909c : SUCCESS in 2m 54s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdd78ef8bccf4678be4c4ad76f14b048 : SUCCESS in 2m 59s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/cf8b4527bffd4f02a71526c14a4fa325 : SUCCESS in 4m 24s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05d237922f84406dab9708f23554aba0 : SUCCESS in 6m 24s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17ea378f16ca4bc8a0f61cb26911c46b : SUCCESS in 6m 55s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81a8d955e0864dc597028aa153cb2969 : SUCCESS in 1h 00m 09s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a0c419e47a5473ca9ed84fcdadbf826 : SUCCESS in 1h 05m 29s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0edd3329d8434d008614cc78a6008e12 : SUCCESS in 50m 47s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2f954760d6345b7af23d68edee2e973 : SUCCESS in 35m 12s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f0c2ca4a8a140f9b26739aa1bf92192 : FAILURE in 1h 04m 17s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbd533219c924ada91c4a419d8a7fdc5 : SUCCESS in 1h 13m 34s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/97eb191377754bfa8fe5e57b12053108 : SUCCESS in 1h 10m 50s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa75b0b76b7e437b8bd01bf884053310 : SUCCESS in 1h 02m 36s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b30620450ff498cbc97d7c65283953b : SUCCESS in 56m 04s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/98ebbe3eff864d09bb6dd00c0e6498b7 : SUCCESS in 58m 26s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ffbb565dc8f44e2b95b7ea8eb9d9835 : SUCCESS in 1h 20m 29s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4da9eb7005364f37907639453c01bd6f : SUCCESS in 1h 23m 41s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d77c77695674b07acfd7302f2166219 : SUCCESS in 52m 16s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1f81caec8ad4a4db6843c4d6593f0f1 : SUCCESS in 51m 47s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fdf1476f4bb4204b3c8fa62fc19b8aa : SUCCESS in 27m 41s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b037ef30a57e4f9d9302fc2fec51708b : SUCCESS in 17m 23s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/212951e22a5e40a6abaa32ce18ae9283 : SUCCESS in 12m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c07c7e431884334a715cfa7d0ae8715 : SUCCESS in 11m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec573782a8764bb3b300acb354c01202 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":2},{"id":"814c753c31d630a5c0ab26a8195c2e752c8f7886","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-09-03 13:29:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9397d1778b36d46d41227bf73654d08260894945","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 13:29:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2b471673c4887065a82b522b801533ca14a5506f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 15:02:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82f0e1a2552a4687bfc6cddc730fb2fc\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ece82e1c7cd444a6a42fdf6a30725a43 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a747ed1f744a470187867a76596d50fe : SUCCESS in 3m 43s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/6450ec69ee4442059b3c4aeb54de64cc : SUCCESS in 7m 39s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60563089fa94417aa4db6172ae3a021b : SUCCESS in 5m 46s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c7f8586712c4b259543ddeee8254b6b : SUCCESS in 1h 06m 24s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2a7db3233644156b9b2e40939ab1096 : SUCCESS in 39m 51s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/059f47d6b4de432192ce2ae5d65a9355 : SUCCESS in 1h 00m 36s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/78700d01d41f43278e76dbc86c9612b0 : SUCCESS in 58m 13s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2b11e64652848ce8eb01b568b0bf1e9 : SUCCESS in 1h 26m 55s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/23a4d3603f52443b93460c9eb8ed4f84 : SUCCESS in 18m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca9dc7a6546d4ec0a9659be0e38ab91d : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"55b14db13c02ea04df6eb10735507a97b5bc4727","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 15:02:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4e72e38c68200d8d34229c8891e2fd7bbd2a6e8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 15:03:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c086b6c7045f418dbaf0abd772b8cb34\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e53a5d1b1c22474585bdeaff7bec40fc : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"91491d4b372beb00c3ff1c108108ecf0e5004e1e","revisions":{"c4f71d360c4eca26f7dc0e3f54adf75403297724":{"kind":"REWORK","_number":1,"created":"2025-09-03 02:00:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/959270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/70/959270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/70/959270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/70/959270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/70/959270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/70/959270/1"}}},"commit":{"parents":[{"commit":"2d71ed5689503962f03e0f737c1d27546089128d","subject":"Merge \"Prepare for voxpupuli-puppet-lint-plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2d71ed5689503962f03e0f737c1d27546089128d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-03 01:58:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-03 02:00:15.000000000","tz":540},"subject":"Remove support for [oslo_messaging_rabbit] heartbeat_in_pthread","message":"Remove support for [oslo_messaging_rabbit] heartbeat_in_pthread\n\nThis feature was introduced to fix missing heartbeat in api services\nrun by external server mechanism (independent from eventlet).\n\nHowever glance api does not use eventlet in this deployment pattern so\nthis parameter has no effect.\n\nBecause this option is deprecated and highly discouraged now due to\nits incompatibility with eventlet, remove it early before the option\nis actually removed.\n\nChange-Id: Ib7a8d4983210c084d83599b4a69623371b09ca9b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c4f71d360c4eca26f7dc0e3f54adf75403297724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c4f71d360c4eca26f7dc0e3f54adf75403297724"}]},"branch":"refs/heads/master"},"91491d4b372beb00c3ff1c108108ecf0e5004e1e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-03 07:30:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/959270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/70/959270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/70/959270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/70/959270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/70/959270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/70/959270/2"}}},"commit":{"parents":[{"commit":"d08f45f0d300902b60ff9419a66d559a6f44b916","subject":"Fix lint error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/d08f45f0d300902b60ff9419a66d559a6f44b916"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-03 01:58:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-03 07:30:02.000000000","tz":540},"subject":"Remove support for [oslo_messaging_rabbit] heartbeat_in_pthread","message":"Remove support for [oslo_messaging_rabbit] heartbeat_in_pthread\n\nThis feature was introduced to fix missing heartbeat in api services\nrun by external server mechanism (independent from eventlet).\n\nHowever glance api does not use eventlet in this deployment pattern so\nthis parameter has no effect.\n\nBecause this option is deprecated and highly discouraged now due to\nits incompatibility with eventlet, remove it early before the option\nis actually removed.\n\nChange-Id: Ib7a8d4983210c084d83599b4a69623371b09ca9b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/91491d4b372beb00c3ff1c108108ecf0e5004e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/91491d4b372beb00c3ff1c108108ecf0e5004e1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
