)]}'
{"id":"openstack%2Fpuppet-glance~962748","triplet_id":"openstack%2Fpuppet-glance~master~I4e11f8f188e57351408637551ba1f337482fa373","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-03 15:18:21.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I4e11f8f188e57351408637551ba1f337482fa373","subject":"Sync supported container/disk format with glance","status":"MERGED","created":"2025-10-02 07:42:15.000000000","updated":"2025-10-03 18:10:22.000000000","submitted":"2025-10-03 18:10:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962748","meta_rev_id":"10d73ba86f84c94132486e780669edd3d3d1ca5e","_number":962748,"virtual_id_number":962748,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-03 18:10:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-10-02 09:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-10-03 13:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-10-03 13:43:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 09:15:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-02 09:23:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-10-03 13:43:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"802463c1e1eec1e9404a2dd5717c55488f048499","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 07:42:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8a76164a7803efdf1050072fca4c361a91e6b6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 09:15:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aad3e4e9a9574e2382a44f1dc6f80781\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91a27714179f4c52a21eb32db890d19a : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e4814c9a6854a42afc54db500b93398 : SUCCESS in 2m 04s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/f3ec3ad115054d82bd6ba80f38c3a2fc : SUCCESS in 4m 59s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57c90ed49c4a4314ba5fb11e0bec38c0 : SUCCESS in 3m 58s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ab56dd56f8948ea987f49529f06f7f6 : SUCCESS in 6m 11s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebc2005ef70e45af84f2167b35a17648 : SUCCESS in 26m 14s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbde95a6536641f5ae2b85f9ca83d52c : SUCCESS in 1h 06m 05s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f8fc53672954790b78c19ddd8d30681 : SUCCESS in 23m 02s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb8613dbb14c473da755e46722095db8 : SUCCESS in 40m 30s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd78da685e664a62b8d6d4e2f31b4f0b : FAILURE in 50m 36s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd2fd5d516944a1a9d0e07366c9134a9 : SUCCESS in 1h 07m 11s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2d3eff6e7004c049e4bf72952c01a6b : SUCCESS in 1h 18m 58s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a1536f3bcf6432cbd5d740ea9184df5 : SUCCESS in 30m 00s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/787dcd619ed34d26a98849b8969e886b : SUCCESS in 28m 41s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c7425fff1ed482a884096417d64d0d3 : SUCCESS in 51m 24s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/253e5af168734281ae813e39e86d411e : SUCCESS in 51m 24s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5133765d470b467aa9cb3f75bf5ced55 : SUCCESS in 1h 28m 29s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/48a515dd490e4d99bac0572465765ae0 : SUCCESS in 29m 22s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9244520713a44278cd7961a9f599af5 : SUCCESS in 55m 21s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbc56ee5afbb4b658315cc55fdc3899e : SUCCESS in 52m 50s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2678eaf5073e4d7f8dd95ca79387aef2 : SUCCESS in 9m 46s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37c7c93dc61e453abe8d0961a709e339 : SUCCESS in 10m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3194cfc726f540a0b80ff78d04015a32 : SUCCESS in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96ef743dae1f53f5d3cc73811998af90464598b1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-02 09:23:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02bb5a0b3229ba2c3491f565c6d8355fd070aedc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-10-03 13:43:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b83b019a2094ed9fdc543ba6670687b076bc79a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 13:44:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4dd7fb59aea39a4963171259f814ff0ee510fae8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:52:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ce1c72a599641de96a13f1a942fd8e4\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9932aab261724c58a552c69de9683d98 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/823fb669457c4a77888722fe8de4f616 : SUCCESS in 3m 11s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/8685f6836afc49bc83cb4610c734e2aa : SUCCESS in 6m 03s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09ac49da48e4467bacba772d3ecd5498 : SUCCESS in 5m 27s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1abd3ccbcd1647019eeb291b14a4b970 : SUCCESS in 1h 04m 32s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6399e8244c01422eb49b07ed9c25d427 : SUCCESS in 43m 57s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51072e8b1c234a649da81293ef1882f1 : SUCCESS in 31m 24s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ded392023f5d45e8a4df35c5f397f428 : SUCCESS in 56m 46s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/272be1c9ed634709b141076239e580f5 : SUCCESS in 45m 12s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a71ee5d18d654ea594822d84de120671 : POST_FAILURE in 10m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5349b558aa6add715256275c3d8283db61ba659f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-03 15:18:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aff5b46576cf6ebfefd2577c5ab8ac316e3d746c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 16:43:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10f708122ea84303a0f6d09077dd2c30\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b0bd735c8ea428f9919de7ba5eb2a19 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca15184f4794407bb3227db1843b413f : SUCCESS in 3m 47s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/8289f54f7fd04c96a6d4786c5cc88ddd : SUCCESS in 4m 22s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2faa61bcc75f44c29e4a2a0e529f39bf : SUCCESS in 3m 35s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b77c2806d1c949bbb6bbc2857e540769 : SUCCESS in 3m 27s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d08aeb8e8074fb18c56f3c14053ce1b : SUCCESS in 59m 35s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eec4c0378f524dfebe70ab7abe47edfc : SUCCESS in 36m 49s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b247552634d4df4a292e013886500b6 : FAILURE in 31m 50s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c2dba7af06c4fc1b568808f9ed71d7c : SUCCESS in 32m 01s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1cc89d32e244d8cb57a68a948cc29fa : FAILURE in 29m 51s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b920127b82c848458b11eaaf02957dbc : SUCCESS in 40m 32s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffdc7d479eda4cafbee53252934f49ea : SUCCESS in 1h 20m 12s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce2d16f8f6ef433bb45cb33630934de0 : SUCCESS in 32m 10s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab11634a599a4ac0909b521f46bd43f4 : SUCCESS in 28m 29s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e16bc1697f1411984f18d02b14de359 : SUCCESS in 54m 40s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d04d7bf11a3484c8e8d375d0ede67d8 : SUCCESS in 1h 04m 48s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f475b18afd041e393376d398eea7b65 : SUCCESS in 1h 13m 10s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f576e607c51b412a824b78d4651528ed : SUCCESS in 43m 10s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e535a1e00b94f2d9ec25fc88bd4ecb1 : SUCCESS in 57m 08s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a759d52197474cf0bb2cfeed3dbe1d26 : SUCCESS in 58m 27s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/6be592a07926447c94cec32a61c55366 : SUCCESS in 15m 38s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66c1b334e3b0454593c86d25d3e7b1ac : SUCCESS in 5m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c26b0269ad943399ec6844e24ad1a82 : SUCCESS in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ddaed731a59c45ad15cee15cbaefc36722cae00c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 16:43:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35bafdd1bb5eb00f04ef650c8af8bbad1d8cfc0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 18:10:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bdbd3ed456841b9b4d958f241564ba4\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36c0d6d2219e4f9096fd0e773a57e445 : SUCCESS in 2m 13s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4250c69fbad49f5948054cac6f64c99 : SUCCESS in 2m 33s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/d9b5d78dc8c84bf08fd55711f3d3d33f : SUCCESS in 4m 19s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/311d7106ac9a4937b8ffcbe19dd4a0bf : SUCCESS in 6m 19s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e02a538079034bcaadaa78271786f28b : SUCCESS in 1h 04m 02s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d042aa8d7fc943d99904bb4af973787c : SUCCESS in 1h 15m 52s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/569f3611f1414277be0325376450c232 : SUCCESS in 59m 14s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6dd934bc336b4c3fa34fea327574c89f : SUCCESS in 54m 53s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbe1cb9f08544a8c967e3a9a80fb94be : SUCCESS in 54m 47s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f35eacbb687347e59c2fb27d70fb7204 : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"10d73ba86f84c94132486e780669edd3d3d1ca5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 18:10:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0584573bb3c529126d66f59bb699f79e57c8ee16","revisions":{"0584573bb3c529126d66f59bb699f79e57c8ee16":{"kind":"REWORK","_number":1,"created":"2025-10-02 07:42:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/962748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/48/962748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/48/962748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/48/962748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/48/962748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/48/962748/1"}}},"commit":{"parents":[{"commit":"2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2","subject":"Merge \"Refactor api service management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 07:41:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 07:42:04.000000000","tz":540},"subject":"Sync supported container/disk format with glance","message":"Sync supported container/disk format with glance\n\nChange-Id: I4e11f8f188e57351408637551ba1f337482fa373\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0584573bb3c529126d66f59bb699f79e57c8ee16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/0584573bb3c529126d66f59bb699f79e57c8ee16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
