)]}'
{"id":"openstack%2Fpuppet-glance~962781","triplet_id":"openstack%2Fpuppet-glance~master~I0253a4fb7b13c0425203dddd5df1766ddc673192","project":"openstack/puppet-glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-03 15:16:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0253a4fb7b13c0425203dddd5df1766ddc673192","subject":"Refactor is_public handling","status":"MERGED","created":"2025-10-02 11:13:57.000000000","updated":"2025-10-03 15:16:53.000000000","submitted":"2025-10-03 15:16:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962781","meta_rev_id":"1c9d39dee7021a0ebf3d41616ceb04c1484189e7","_number":962781,"virtual_id_number":962781,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-03 15:16:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-10-02 21:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-10-03 13:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-10-03 13:43:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 12:02:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-02 21:06:53.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-10-03 13:43:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"b783f99cbd677062fb6651a8175c87ebbd143a59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 11:13:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88d3554664c5db71e24a2c471c69f9b4824cc2f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 11:45:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c1e526f59ad7d5e38d44b5fd5b627cdce99f14fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 12:02:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c0d988cd9c3434e939dca32e57d5aa7\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a23006c8f4f429fa1a26da062da4afd : SUCCESS in 2m 41s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcece9fca32d486e9747a36fa88d9b55 : SUCCESS in 2m 49s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/100a0fea994945758f4c14cd51db406e : FAILURE in 5m 15s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fa4ea93961a4fe0869d5de708e0d246 : FAILURE in 5m 58s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d20ec0f95cbb481c8a6b452f45a27f60 : FAILURE in 6m 10s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33eb339ad42d4cc3bbc20e1d8b065daa : FAILURE in 4m 57s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a786f540d6f945a09f53871016af330a : FAILURE in 5m 17s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ca8ef8dbf444a98b318febbb25b652f : FAILURE in 4m 08s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1322b501ecd46008e178904828fab0e : FAILURE in 5m 52s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23908dafdea6442fbbfb9a98304ee411 : FAILURE in 5m 22s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1148bbf5b5544664966d28ae3dd17b6a : FAILURE in 9m 59s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7760941fcd84e8c8dc14a855c370e83 : FAILURE in 9m 19s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5076fc6e0f22451694617f286eb3541d : FAILURE in 8m 10s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d5216359277442b8189eb684c15301e : FAILURE in 10m 35s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e04689271dba4e9f9ec0c81ec7b149a2 : FAILURE in 9m 25s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cf26f23cec24f49add56a29e1e44213 : FAILURE in 9m 54s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bc36807764d40da930410cc88597fa5 : FAILURE in 9m 05s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e8a52556e354c4fa3e00945f0b411ae : FAILURE in 10m 20s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ea9d4e03fe4455e857784326de6316a : FAILURE in 8m 58s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed19e6e524024f5bb98c2de232dac5bb : FAILURE in 9m 56s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/db372ecb3fa744eda650209bdab0216b : FAILURE in 7m 46s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ceee9e51d0b444e8fb85904c0df6ddc : FAILURE in 4m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d848b525db12492caec01787b71550bb : FAILURE in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ebe3384e374212752892e070e73dd9bb67bc503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 12:29:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2ebe1eeb7db7739748d0161f7f4b6c6175dee15f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 12:29:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"12023baf12d6caeefafd8f7186657538bf76e9cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 12:47:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c0c443068168ccc855c5fc6080e2f63ce2f8e9d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 12:49:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a38e0c3f9f5128570f935081d1c9a37125c26278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 14:14:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7be90e1eef94654ab4827f5076edec2\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92aef5105b864ac1a04c17144d55c8aa : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/965cbfa2175342b79b6bbe18e04e84b6 : SUCCESS in 3m 32s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/968de259806c4fb9bf517c6e22bb0425 : SUCCESS in 5m 41s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ef60e3d08674cc7ba07ca083c831ee8 : SUCCESS in 6m 41s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ecddb35836544ee951ba39fb95ef560 : SUCCESS in 6m 51s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e25d0cd941b94659aa287b78c0ba3010 : SUCCESS in 44m 37s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5981a8233ad44e28161e4acfd242385 : SUCCESS in 1h 05m 59s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c643904445f9437393ea289258e379a8 : SUCCESS in 50m 16s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18b069502b074082b4003e2e267bde23 : SUCCESS in 43m 12s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bee2e0f6189b49b49810d7cd17da3cc3 : FAILURE in 24m 57s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/10de006f147d41c7a612f28999113449 : SUCCESS in 40m 53s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51fe0d888fb541bab9272d264a4c1e7b : SUCCESS in 44m 51s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c327aabd6182463e9b53daf7bdd9ee16 : SUCCESS in 27m 07s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b73cf7f4dd24a65af02f99ba946a663 : SUCCESS in 53m 21s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8b6ecf542f64cd1b2939fa4de55df25 : SUCCESS in 28m 13s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/438d196c7e6e4f32b4d9a6d89d3d50c6 : SUCCESS in 1h 00m 10s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff55effca9c24fdcb06c93cec6ad635c : SUCCESS in 1h 19m 05s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/77d18c694e4a45ea90c5338d461e4a73 : SUCCESS in 27m 31s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9317d4bc83cb4e81bcef366cb441768f : SUCCESS in 44m 21s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2376dc136874cd9ae73210e9f385159 : SUCCESS in 58m 58s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bc8b86d04bd475ab1c427d54cd38491 : SUCCESS in 8m 25s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63fcd28ac0234ddca80fef369651335f : SUCCESS in 11m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28e277c5214f421e9515c2e50b10dd01 : SUCCESS in 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5e30fb479c780e8093999847781f1e737d0bfea4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-02 21:06:53.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"195dc006f3b1480d6a72ca3667cfa5791e53b380","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-10-03 13:43:24.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"314aa9fadedf05a3803d7f4570c464104dfcb9ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 13:44:07.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"61aea80cc855465d8de93b1d5454f5f4f03a530d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 15:16:53.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/318a7154303749dfaae839ababa734f1\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1777b4bd0f048cb86d26446b1e1415f : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6e8cb7740dc4c9a8eb57bafbf9c0376 : SUCCESS in 2m 12s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/c16d7cb618f041cc8c463903da93195e : SUCCESS in 6m 35s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68fa449724144e139ac67d6c3f0bde71 : SUCCESS in 3m 26s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e66f48ea8354aa28adc60f6f279f5f7 : SUCCESS in 1h 05m 36s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd3169e3bc574a5aa6cc87f8ec0902a9 : SUCCESS in 1h 17m 19s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5bf6a9ef8ee4c62bb5dbdfb1ca04789 : SUCCESS in 46m 29s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e85ce3d90305446e8b68d830a1749541 : SUCCESS in 45m 59s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9fe4bb30a6148f6a463824749f0c6db : SUCCESS in 46m 59s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ee66b655f714433a771a252dd501a74 : SUCCESS in 17m 27s","accounts_in_message":[],"_revision_number":6},{"id":"1c9d39dee7021a0ebf3d41616ceb04c1484189e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 15:16:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"93d697dec5daecb88422ef3e119a6fb827c97701","revisions":{"680020cee63f1f2d5850518f3b267eb76c30926d":{"kind":"REWORK","_number":1,"created":"2025-10-02 11:13:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/962781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/962781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/1"}}},"commit":{"parents":[{"commit":"2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2","subject":"Merge \"Refactor api service management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 11:13:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 11:13:49.000000000","tz":540},"subject":"Drop redundant bool_to_sym function","message":"Drop redundant bool_to_sym function\n\nIt\u0027s used only in a single place.\n\nChange-Id: I0253a4fb7b13c0425203dddd5df1766ddc673192\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/680020cee63f1f2d5850518f3b267eb76c30926d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/680020cee63f1f2d5850518f3b267eb76c30926d"}]},"branch":"refs/heads/master"},"6a08fd6c6bae2169c9ab526897cea47fce356419":{"kind":"REWORK","_number":2,"created":"2025-10-02 11:45:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/962781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/962781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/2"}}},"commit":{"parents":[{"commit":"2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2","subject":"Merge \"Refactor api service management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 11:13:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 11:44:49.000000000","tz":540},"subject":"Refactor is_public handling","message":"Refactor is_public handling\n\nConsistently use sym instead of mixing string/boolean.\n\nChange-Id: I0253a4fb7b13c0425203dddd5df1766ddc673192\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6a08fd6c6bae2169c9ab526897cea47fce356419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/6a08fd6c6bae2169c9ab526897cea47fce356419"}]},"branch":"refs/heads/master"},"cbfdaaf8e7ee3e9bfd0e79be70828634b07af3f2":{"kind":"REWORK","_number":3,"created":"2025-10-02 12:29:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/962781/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/962781/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/3"}}},"commit":{"parents":[{"commit":"2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2","subject":"Merge \"Refactor api service management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 11:13:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 12:29:07.000000000","tz":540},"subject":"Refactor is_public handling","message":"Refactor is_public handling\n\nConsistently use sym instead of mixing string/boolean.\n\nChange-Id: I0253a4fb7b13c0425203dddd5df1766ddc673192\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cbfdaaf8e7ee3e9bfd0e79be70828634b07af3f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/cbfdaaf8e7ee3e9bfd0e79be70828634b07af3f2"}]},"branch":"refs/heads/master"},"c632322f6de83af259a05578d1b885c7b336b9f7":{"kind":"REWORK","_number":4,"created":"2025-10-02 12:29:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/962781/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/962781/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/4"}}},"commit":{"parents":[{"commit":"2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2","subject":"Merge \"Refactor api service management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 11:13:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 12:29:47.000000000","tz":540},"subject":"Refactor is_public handling","message":"Refactor is_public handling\n\nConsistently use sym instead of mixing string/boolean.\n\nChange-Id: I0253a4fb7b13c0425203dddd5df1766ddc673192\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c632322f6de83af259a05578d1b885c7b336b9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/c632322f6de83af259a05578d1b885c7b336b9f7"}]},"branch":"refs/heads/master"},"dd6840d946f252c7a17f93ac22be7cc5229bb4f1":{"kind":"REWORK","_number":5,"created":"2025-10-02 12:47:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/962781/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/962781/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/5"}}},"commit":{"parents":[{"commit":"2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2","subject":"Merge \"Refactor api service management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 11:13:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 12:47:15.000000000","tz":540},"subject":"Refactor is_public handling","message":"Refactor is_public handling\n\nConsistently use sym instead of mixing string/boolean.\n\nChange-Id: I0253a4fb7b13c0425203dddd5df1766ddc673192\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/dd6840d946f252c7a17f93ac22be7cc5229bb4f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/dd6840d946f252c7a17f93ac22be7cc5229bb4f1"}]},"branch":"refs/heads/master"},"93d697dec5daecb88422ef3e119a6fb827c97701":{"kind":"REWORK","_number":6,"created":"2025-10-02 12:49:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/962781/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/81/962781/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/81/962781/6"}}},"commit":{"parents":[{"commit":"2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2","subject":"Merge \"Refactor api service management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/2d6712e2adb4efef97a6a0cf3c0357f7f7b82cd2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 11:13:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 12:49:30.000000000","tz":540},"subject":"Refactor is_public handling","message":"Refactor is_public handling\n\nConsistently use sym instead of mixing string/boolean.\n\nChange-Id: I0253a4fb7b13c0425203dddd5df1766ddc673192\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/93d697dec5daecb88422ef3e119a6fb827c97701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/93d697dec5daecb88422ef3e119a6fb827c97701"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
