)]}'
{"id":"openstack%2Fpuppet-glance~993226","triplet_id":"openstack%2Fpuppet-glance~master~Ifd8ff40f14cddcb44a54a8c01b9c3d021ef0c1aa","project":"openstack/puppet-glance","branch":"master","topic":"remove-postgresql","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-04 17:04:34.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ifd8ff40f14cddcb44a54a8c01b9c3d021ef0c1aa","subject":"Remove PostgreSQL support","status":"MERGED","created":"2026-06-13 14:53:21.000000000","updated":"2026-07-04 18:49:20.000000000","submitted":"2026-07-04 18:48:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993226-remove-postgresql","meta_rev_id":"8242b47cd6d8392638ffb2aa97e8a866e9b53930","_number":993226,"virtual_id_number":993226,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-04 18:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-07-04 17:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-06-30 06:15:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-07-04 17:04:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-13 16:38:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-30 06:15:24.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"322ef28e1c5b382d770357f38c9d61631f3343af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-13 14:53:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"597353309685d9512e8a171b701a328df8b40dfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-13 15:00:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ea00e874281ff820228c0c44f06eac0808977c37","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-13 15:00:51.000000000","message":"Topic set to remove-postgresql","accounts_in_message":[],"_revision_number":2},{"id":"151d736d9fdee1575a31ed1c4658312ce51dcc39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 16:38:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/834d128bcba1446f98ce67cb7804da35\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bc5b65a69cc4437a8a1a7dec654a742 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/476b003be55c4c7db89133a218351b5e : SUCCESS in 3m 18s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/b65eb7040ab746ed8c1980597519b037 : SUCCESS in 9m 13s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2cca77a4d1c4206848e79d3da09dbd4 : SUCCESS in 9m 29s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c471a068d0d149caa35385299934667b : SUCCESS in 8m 19s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b62370e5bfd471fbb8a55d056fa1b07 : SUCCESS in 1h 02m 30s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c31041489dc4959aab591af88e835c7 : FAILURE in 1h 01m 49s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a568b8ed2f484417845bbfab869f8a1d : FAILURE in 1h 14m 38s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/307946310de8438a8cf62d93976bcf95 : SUCCESS in 50m 33s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8ad39e2fe9c437b90e3073b864596b0 : FAILURE in 52m 53s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a369e53b80624c2baf43cc2ff8cc38bf : SUCCESS in 36m 42s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b74e388aa9a4d50a5aed665af782fba : SUCCESS in 1h 18m 35s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/657ccc5b14994250a59ecaf174b74d91 : SUCCESS in 1h 01m 31s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5c4a0fa520545f7b84eee847df4b9a7 : SUCCESS in 57m 06s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/06e69915704a4d9bb2e096e1303322d6 : SUCCESS in 44m 43s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54b192591267426a87d8906763452cf6 : SUCCESS in 59m 36s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cf7ab9666a34da6a37153fe3c0dbd5c : SUCCESS in 1h 15m 54s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c87d5eca8e2d40399b110634ee500b64 : SUCCESS in 33m 27s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/68fe78d19c464c629b1c465e1fcbc921 : SUCCESS in 55m 10s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6665407e6db546148b528bfc83f28d23 : SUCCESS in 1h 00m 05s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e96d290fbf64b38b89037147117d6e5 : SUCCESS in 18m 42s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/233d2db9e79140f287f2fd354e4a8340 : SUCCESS in 13m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9888511554646889986c78755a5349c : SUCCESS in 14m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6575205af4e14a4a8b352887ae870a36 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d65490cc504b9b56b57e7699f0b3276938476a60","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 06:15:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ef2d5a96fb979ef1b791fd4ae2978427fb12a73b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-04 17:04:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c44f152472b63e4dd47a741fb4353cca94b71dae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 17:06:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4d19d2da615c58cd69ad8a6e48524abbe871f0b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 18:48:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/406cd40122b34ec594a0b3a3e433bf02\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77c098442577405cafb6a32fbab0ac9e : SUCCESS in 2m 40s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c87fb92f029445eb0bb7950480aaa9f : SUCCESS in 2m 32s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/179b0ecbda884334935467a90f88a055 : SUCCESS in 16m 47s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2ae079888be476a850b35b2c2305597 : SUCCESS in 7m 09s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a6fda9ad1bf4e168cbbc62021ff2c1c : SUCCESS in 1h 03m 31s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd2d1029955543fbbf9bd0d10c0f8b9a : SUCCESS in 1h 17m 58s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/47214be7550d4b78a5c6dbfc10029027 : SUCCESS in 51m 58s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/249cb7a5b60d4de0b541998faa0e9e5b : SUCCESS in 46m 26s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd601ff11ecc4bdd8cbc7f4fcf318471 : SUCCESS in 45m 49s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/62c22038ad194451b845af8d9ea39047 : SUCCESS in 19m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39429a4a33774523826fad794a57189a : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"4a5042c77af6a9f9b97f23f17b486e75e1a57c43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 18:48:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8242b47cd6d8392638ffb2aa97e8a866e9b53930","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 18:49:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876dbecd0b7442549470c4dc425d636a\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e428823b28bb4e66a045f3f7a4ee1b52 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9840cfc4ad8174f61699721e3e8206812e101524","revisions":{"9271cd97552e39fb619b35b8fd0a75833b3e98f7":{"kind":"REWORK","_number":1,"created":"2026-06-13 14:53:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/993226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/26/993226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/993226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/993226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/993226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/26/993226/1"}}},"commit":{"parents":[{"commit":"1df657ffd1ddc0beaf4062451bad7dc7a64933c2","subject":"Bump upper version of puppetlabs-stdlib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1df657ffd1ddc0beaf4062451bad7dc7a64933c2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-13 14:53:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-13 14:53:07.000000000","tz":540},"subject":"Remove PostgreSQL support","message":"Remove PostgreSQL support\n\nIt was deprecated a few cycles back[1], due to neutron which already\nremoved PostgreSQL support.\n\n[1] 0c4812f8815812a0e938089239f5a4964f8e37c4\n\nChange-Id: Ifd8ff40f14cddcb44a54a8c01b9c3d021ef0c1aa\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9271cd97552e39fb619b35b8fd0a75833b3e98f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9271cd97552e39fb619b35b8fd0a75833b3e98f7"}]},"branch":"refs/heads/master"},"9840cfc4ad8174f61699721e3e8206812e101524":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-13 15:00:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/993226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glance","ref":"refs/changes/26/993226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/993226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/993226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glance refs/changes/26/993226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glance refs/changes/26/993226/2"}}},"commit":{"parents":[{"commit":"1df657ffd1ddc0beaf4062451bad7dc7a64933c2","subject":"Bump upper version of puppetlabs-stdlib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/1df657ffd1ddc0beaf4062451bad7dc7a64933c2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-13 14:53:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-13 15:00:01.000000000","tz":0},"subject":"Remove PostgreSQL support","message":"Remove PostgreSQL support\n\nIt was deprecated during the previous cycle[1], due to neutron which\nalready removed PostgreSQL support.\n\n[1] 0c4812f8815812a0e938089239f5a4964f8e37c4\n\nChange-Id: Ifd8ff40f14cddcb44a54a8c01b9c3d021ef0c1aa\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9840cfc4ad8174f61699721e3e8206812e101524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glance/commit/9840cfc4ad8174f61699721e3e8206812e101524"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
