)]}'
{"id":"openstack%2Fpuppet-glare~654197","triplet_id":"openstack%2Fpuppet-glare~master~Ie713347962a0b60bb1d90bb9c4a7eb5c668af6a6","project":"openstack/puppet-glare","branch":"master","topic":"move-to-opendev","hashtags":[],"change_id":"Ie713347962a0b60bb1d90bb9c4a7eb5c668af6a6","subject":"Move to opendev","status":"MERGED","created":"2019-04-21 06:45:40.000000000","updated":"2019-04-22 17:17:01.000000000","submitted":"2019-04-22 17:17:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654197-1555953421510-a663f06a","meta_rev_id":"6e44986cb2fe80619e12a25545cdfe7888dee968","_number":654197,"virtual_id_number":654197,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-22 17:17:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-04-22 13:27:28.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-22 16:55:44.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-04-22 16:55:44.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 13:27:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-22 16:55:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-22 17:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50034511ab9fcdad8eec54b7dc48cc1e180a8496","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 06:45:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d89fc249023c9cdb2384ed4ae7fa9486015ff83a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 07:12:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/97/654197/1/check/puppet-openstack-lint-ubuntu-bionic/42fe748/ : SUCCESS in 6m 31s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/97/654197/1/check/puppet-openstack-syntax-4-ubuntu-bionic/8f9c536/ : SUCCESS in 6m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/97/654197/1/check/puppet-openstack-syntax-5-ubuntu-bionic/aa14f44/ : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/97/654197/1/check/puppet-openstack-syntax-6-ubuntu-bionic/5a63744/ : SUCCESS in 6m 35s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/97/654197/1/check/puppet-openstack-unit-4.8-centos-7/e36ce5b/ : SUCCESS in 16m 50s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/97/654197/1/check/puppet-openstack-unit-5.5-centos-7/c131b3b/ : SUCCESS in 16m 47s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/97/654197/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/d1b3d07/ : SUCCESS in 7m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/97/654197/1/check/puppet-openstack-unit-latest-ubuntu-bionic/8b89de6/ : FAILURE in 14m 49s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/97/654197/1/check/puppet-openstack-beaker-centos-7/dabf21a/ : SUCCESS in 9m 08s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/97/654197/1/check/puppet-openstack-beaker-ubuntu-bionic/e9b5260/ : SUCCESS in 9m 00s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/97/654197/1/check/build-openstack-releasenotes/9c7f58f/html/ : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"197add1b2068106b09d0961d43d6fe39bfb0265a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-22 13:27:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8c93840b6d91f475645de4641b6126a8ad0b6c4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-22 16:55:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2fd966f3efa29a7124e51da3790126a4b3e0328","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:56:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6c252bee898e0b5339e83d31a58210f9cdc6079","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:17:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b8b8a6fb01b4a65de07b8866759f1bbac26ab1f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:17:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/97/654197/1/gate/puppet-openstack-lint-ubuntu-bionic/d17cddc/ : SUCCESS in 4m 54s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/97/654197/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/b1a0dcb/ : SUCCESS in 4m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/97/654197/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/b1aa90a/ : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/97/654197/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/29e9cf7/ : SUCCESS in 4m 06s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/97/654197/1/gate/puppet-openstack-unit-4.8-centos-7/0ecebfd/ : SUCCESS in 10m 46s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/97/654197/1/gate/puppet-openstack-unit-5.5-centos-7/c9121a1/ : SUCCESS in 10m 51s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/97/654197/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/e60feaf/ : SUCCESS in 7m 37s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/97/654197/1/gate/puppet-openstack-beaker-centos-7/781cda9/ : SUCCESS in 9m 04s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/97/654197/1/gate/puppet-openstack-beaker-ubuntu-bionic/dc8cd0c/ : SUCCESS in 9m 31s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/97/654197/1/gate/build-openstack-releasenotes/1b6ebfd/html/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4219400de72b3f9a589383eb21367d2ea037391d","revisions":{"4219400de72b3f9a589383eb21367d2ea037391d":{"kind":"REWORK","_number":1,"created":"2019-04-21 06:45:40.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/97/654197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/97/654197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/97/654197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/97/654197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/97/654197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/97/654197/1"}}},"commit":{"parents":[{"commit":"9267b66a25200e9b982d6aaa9677b8bce59820e0","subject":"Fix Repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/9267b66a25200e9b982d6aaa9677b8bce59820e0"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:48.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:48.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: Ie713347962a0b60bb1d90bb9c4a7eb5c668af6a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/4219400de72b3f9a589383eb21367d2ea037391d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/4219400de72b3f9a589383eb21367d2ea037391d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
