)]}'
{"id":"openstack%2Fpuppet-glare~663186","triplet_id":"openstack%2Fpuppet-glare~master~Ib00c48750b86976a1d00b3ed3e451497961b94d3","project":"openstack/puppet-glare","branch":"master","topic":"prep-release-train-1","hashtags":[],"change_id":"Ib00c48750b86976a1d00b3ed3e451497961b94d3","subject":"Prepare Train M2","status":"MERGED","created":"2019-06-05 01:42:52.000000000","updated":"2019-06-10 13:34:36.000000000","submitted":"2019-06-10 13:34:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663186-1560173676887-0eece10d","meta_rev_id":"eb7fb4c5b662853c7237755552c27e278f1810f8","_number":663186,"virtual_id_number":663186,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-10 13:34:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-06-10 13:16:58.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-06-10 13:16:58.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 13:16:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-10 13:34:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29b38a0560adde3953eaf19052e1986c65edaa7d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-05 01:42:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9458dbceafe822777a8ebb7e2942868dcc2d9fe3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:09:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/86/663186/1/check/build-openstack-puppet-tarball/37eccf8/ : SUCCESS in 3m 33s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/86/663186/1/check/puppet-openstack-lint-ubuntu-bionic/da8307e/ : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6df7079870890d18c9f8564ed7ee0bfbdd221db6","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-10 02:09:06.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1d18042824e9b1ce466e63cb5a343c68d301cba4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 02:21:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/86/663186/2/check/build-openstack-puppet-tarball/153ae6d/ : SUCCESS in 2m 19s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/86/663186/2/check/puppet-openstack-lint-ubuntu-bionic/981e936/ : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":2},{"id":"dd960e650c2730687d2ae2bbf49071e445a6950e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:16:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a31b569eb6362496a491a026f35a42c5bd803b90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:17:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a1c7c6134231494912e58201f39dc159bee4384","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:34:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c8cd4e430107e6af22b7782441de55c072f6fe8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:34:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/86/663186/2/gate/build-openstack-puppet-tarball/9643869/ : SUCCESS in 2m 16s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/86/663186/2/gate/puppet-openstack-lint-ubuntu-bionic/dd6563d/ : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f1a5ae071dc5f3c9866c2f5bc9883a4f839f025","revisions":{"085e1e45b4e0832fba580e46014ddb8b7caf8a54":{"kind":"REWORK","_number":1,"created":"2019-06-05 01:42:52.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/86/663186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/86/663186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/86/663186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/86/663186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/86/663186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/86/663186/1"}}},"commit":{"parents":[{"commit":"625ef78c6eae211350401b6c13214580372faffe","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/625ef78c6eae211350401b6c13214580372faffe"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:18.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:18.000000000","tz":480},"subject":"Prepare Train M1","message":"Prepare Train M1\n\nUpdate the version metadata for Train milestone 1\n\nChange-Id: Ib00c48750b86976a1d00b3ed3e451497961b94d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/085e1e45b4e0832fba580e46014ddb8b7caf8a54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/085e1e45b4e0832fba580e46014ddb8b7caf8a54"}]},"branch":"refs/heads/master"},"9f1a5ae071dc5f3c9866c2f5bc9883a4f839f025":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-10 02:09:06.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/86/663186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/86/663186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/86/663186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/86/663186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/86/663186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/86/663186/2"}}},"commit":{"parents":[{"commit":"625ef78c6eae211350401b6c13214580372faffe","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/625ef78c6eae211350401b6c13214580372faffe"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:18.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-10 02:02:57.000000000","tz":480},"subject":"Prepare Train M2","message":"Prepare Train M2\n\nUpdate the version metadata for Train milestone 2\n\nChange-Id: Ib00c48750b86976a1d00b3ed3e451497961b94d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/9f1a5ae071dc5f3c9866c2f5bc9883a4f839f025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/9f1a5ae071dc5f3c9866c2f5bc9883a4f839f025"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
