)]}'
{"id":"openstack%2Fpuppet-glare~665789","triplet_id":"openstack%2Fpuppet-glare~master~I47f459f9223d6b35228474278e2f943a3e416b72","project":"openstack/puppet-glare","branch":"master","topic":"665789","hashtags":[],"change_id":"I47f459f9223d6b35228474278e2f943a3e416b72","subject":"assert_private() params and deps classes","status":"ABANDONED","created":"2019-06-17 23:18:37.000000000","updated":"2020-09-18 22:56:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a8090ae5af3e2e6a8bcfca568fb4d2ebf9b223a8","_number":665789,"virtual_id_number":665789,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-18 00:10:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 00:10:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"377789091a6581b6d1bad69e3ea9a337150a1fa4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:18:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"208dcfd16cefdfde93e30fab005a4efb7db17e94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 23:36:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/89/665789/1/check/puppet-openstack-lint-ubuntu-bionic/3851a3e/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/89/665789/1/check/puppet-openstack-syntax-5-ubuntu-bionic/2ba0fe8/ : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/89/665789/1/check/puppet-openstack-syntax-6-ubuntu-bionic/4b47924/ : SUCCESS in 4m 11s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/89/665789/1/check/puppet-openstack-unit-5.5-centos-7/beef612/ : SUCCESS in 10m 07s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/89/665789/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/6bcc01f/ : SUCCESS in 7m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/89/665789/1/check/puppet-openstack-unit-latest-ubuntu-bionic/b734f69/ : SUCCESS in 7m 05s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/89/665789/1/check/puppet-openstack-beaker-centos-7/73c1a5b/ : SUCCESS in 9m 04s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/89/665789/1/check/puppet-openstack-beaker-ubuntu-bionic/9225538/ : SUCCESS in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c578bae4d55d041be808fdd66f3ce480a628754","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:42:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f54732f9504ec8cc95e5444f222ed6255da8becd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:10:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/89/665789/2/check/puppet-openstack-lint-ubuntu-bionic/83eb124/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/89/665789/2/check/puppet-openstack-syntax-5-ubuntu-bionic/5dce581/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/89/665789/2/check/puppet-openstack-syntax-6-ubuntu-bionic/ffbc061/ : SUCCESS in 4m 13s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/89/665789/2/check/puppet-openstack-unit-5.5-centos-7/631911c/ : SUCCESS in 10m 18s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/89/665789/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/85943ff/ : SUCCESS in 11m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/89/665789/2/check/puppet-openstack-unit-latest-ubuntu-bionic/c02d2ae/ : SUCCESS in 7m 33s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/89/665789/2/check/puppet-openstack-beaker-centos-7/99fa571/ : SUCCESS in 10m 01s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/89/665789/2/check/puppet-openstack-beaker-ubuntu-bionic/b612928/ : SUCCESS in 7m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e3cb84413844e25f34aee3223d244b9605274c2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-18 22:56:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0114202fff65df08ac656764383994caa176d528","revisions":{"b2244bf4e45efc0f847e8d5638419da9ba9601e9":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:18:37.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/89/665789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/89/665789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/665789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/665789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/665789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/89/665789/1"}}},"commit":{"parents":[{"commit":"9f1a5ae071dc5f3c9866c2f5bc9883a4f839f025","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/9f1a5ae071dc5f3c9866c2f5bc9883a4f839f025"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:18:29.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:18:29.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I47f459f9223d6b35228474278e2f943a3e416b72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/b2244bf4e45efc0f847e8d5638419da9ba9601e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/b2244bf4e45efc0f847e8d5638419da9ba9601e9"}]},"branch":"refs/heads/master"},"0114202fff65df08ac656764383994caa176d528":{"kind":"REWORK","_number":2,"created":"2019-06-17 23:42:27.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/89/665789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/89/665789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/665789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/665789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/665789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/89/665789/2"}}},"commit":{"parents":[{"commit":"9f1a5ae071dc5f3c9866c2f5bc9883a4f839f025","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/9f1a5ae071dc5f3c9866c2f5bc9883a4f839f025"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:18:29.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:42:21.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I47f459f9223d6b35228474278e2f943a3e416b72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/0114202fff65df08ac656764383994caa176d528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/0114202fff65df08ac656764383994caa176d528"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
