)]}'
{"id":"openstack%2Fpuppet-glare~707970","triplet_id":"openstack%2Fpuppet-glare~master~Ica4c8766cdd7f088e3ab41c73421310957cbbf9b","project":"openstack/puppet-glare","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Ica4c8766cdd7f088e3ab41c73421310957cbbf9b","subject":"[ussuri][goal] Cleanup for python 2.7 drop","status":"MERGED","created":"2020-02-16 00:59:00.000000000","updated":"2020-02-18 03:15:13.000000000","submitted":"2020-02-18 03:13:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707970-1581995616303-4edda447","meta_rev_id":"b8f667b45ef5a1364dfc985fb24c4323fb15eda6","_number":707970,"virtual_id_number":707970,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 03:13:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-02-18 02:46:28.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-02-17 20:25:21.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-02-18 02:46:28.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 20:25:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-18 02:46:28.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-02-18 03:13:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c4b1ab423e2fe9aafd8bf1bdd97678d0161fe8c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-16 00:59:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b71f383945f36bf71eebe59e72cc893b2c468d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 01:24:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a42cfbb7d0c4dea9bc0fe67ecbd0b15 : SUCCESS in 4m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c35dd3716b194f26b212835d153b75e4 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5cc40d978c74522ab3f1ed76fb1baf6 : SUCCESS in 4m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bf33a617e8ca4152afca18c86923804c : SUCCESS in 9m 38s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f4d49ae631f4b44948d7be70ad42a79 : SUCCESS in 7m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f42d968bb3545329d072d13146b0632 : SUCCESS in 8m 17s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3fa8af1918b3424c8ee4dd4dc0193726 : SUCCESS in 16m 16s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/054acc4b2b5e4466b85031daa0f15ef9 : SUCCESS in 7m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c0197b3f5a14087b6ace6a846322e56 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6b5f40c1bfa06bcd15aee49166741a99d1020b38","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-17 20:25:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc755fe6369290793a855220a9c48bf59cf66b37","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-02-18 02:46:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac8a4ec622d60f49ee23a5beeaf56cb3b6b42186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 02:46:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3780a2ad8ad4abcc5cf6f9ed47ec912854ef02f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 03:13:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"15eb72f26781e2bb25feec4edb784456aec94479","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 03:13:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c601d30a6f04b87bbe2a535901c76e9 : SUCCESS in 6m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/521eca1b6e594b2e8d29db128966a629 : SUCCESS in 6m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11a62b9a0c7c4acda70acfe447e47965 : SUCCESS in 7m 42s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/27b5154c2e6c417fa8c547b49949c105 : SUCCESS in 20m 26s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc337b1858124343bf9420d46968b9c8 : SUCCESS in 14m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ee7f5315705a454689e1ef64ec300cd5 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a42dd85bea8e4fa68c97cb4ded08bd0d : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"5e62bef0469a323791eef3f70097788f70a34bc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 03:15:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a701ab4c6a94f1686b3ce3c1e41cbdc : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d8df2147168b6a3820fd8f20e5cc6d8264968ed","revisions":{"3d8df2147168b6a3820fd8f20e5cc6d8264968ed":{"kind":"REWORK","_number":1,"created":"2020-02-16 00:59:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/70/707970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/70/707970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/70/707970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/70/707970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/70/707970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/70/707970/1"}}},"commit":{"parents":[{"commit":"46fb3150e69034cf98abd0b3d4d320caf9e757b0","subject":"Prepare Ussuri M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/46fb3150e69034cf98abd0b3d4d320caf9e757b0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-16 00:58:21.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-16 00:58:59.000000000","tz":-360},"subject":"[ussuri][goal] Cleanup for python 2.7 drop","message":"[ussuri][goal] Cleanup for python 2.7 drop\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\npuppet-* repos have to cleanup the requirement and tox\nfor py2.7 drop.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Ica4c8766cdd7f088e3ab41c73421310957cbbf9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/3d8df2147168b6a3820fd8f20e5cc6d8264968ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/3d8df2147168b6a3820fd8f20e5cc6d8264968ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
