)]}'
{"id":"openstack%2Fpuppet-glare~714368","triplet_id":"openstack%2Fpuppet-glare~master~I5beee515ab6ceb5df7ef54fd35189a5c72ba242f","project":"openstack/puppet-glare","branch":"master","topic":"remove_idle_timeout_option","hashtags":[],"change_id":"I5beee515ab6ceb5df7ef54fd35189a5c72ba242f","subject":"Remove idle_timeout option","status":"MERGED","created":"2020-03-23 06:47:10.000000000","updated":"2020-03-23 23:37:47.000000000","submitted":"2020-03-23 23:35:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714368-1585006538667-61dee02a","meta_rev_id":"ca27bb7a92e7c7c4644c4c2109a2cc12f5044abb","_number":714368,"virtual_id_number":714368,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 23:35:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-23 13:01:00.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-23 23:19:02.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-23 23:19:02.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 13:01:00.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-23 23:19:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-23 23:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a0c8a26681ebb5123771211bd8dcc00b4331124","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-03-23 06:47:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8a29257bdba8b8c53a0a915c9a227a3ed5aa82b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 07:17:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3d3f7219ccb435893153c8e77f171f0 : SUCCESS in 4m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7b503a235dc4d90a5dc2a4c538264e3 : SUCCESS in 5m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdd06080d70841508ba2b9233a0ba69e : SUCCESS in 4m 40s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e28d7936288848aa896150dc45026a3f : SUCCESS in 17m 32s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/834155e1e30d4b60a6a7d05c0b9da22c : SUCCESS in 15m 28s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/7ee409cac4134efa8a1d2c44a17f31d5 : SUCCESS in 12m 11s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afc6b55b5c34456d84deae6aa5ced634 : SUCCESS in 11m 10s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddc6d76ffdb74ca28b31be55ff05dd02 : SUCCESS in 8m 45s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a27ab48a474040ecb611594ef61f932a : SUCCESS in 10m 46s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b510910bf5b94583a9affa75504fc4ab : FAILURE in 7m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcc551dc08c24e27850639c8f6ab9869 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d129693db5ed7ff8e0013da3004e03df3bb6c392","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-23 13:01:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc3b10b760cbff80fb5d57201f6defaa70c46633","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-23 23:19:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1efd9a9647e8264aa2b157110975c51c102d1bd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 23:19:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6aa1317487572a09c4b7d32f574238b83287912e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 23:35:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"defd464469067530f063c7d97913ebd1598d8861","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 23:35:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3983d381b6c742fdb55120f073a83a3e : SUCCESS in 5m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09d2656aa9d246f280e93d512e53659c : SUCCESS in 6m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37b04560a9e84a5b8a5942745d7ab60b : SUCCESS in 5m 42s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d5093b0c5ec44d6286f6335c76f17cd8 : SUCCESS in 12m 52s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/640c5ff2bbb944e3a0bc836ca9bbc0c7 : SUCCESS in 9m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/02a043f8bb2f4915955a3b22e7dc8ca3 : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8201dfdba32343809ac69265cd2107b7 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"054b8eb12ba05856bb6b236ba0c5af558dfb1374","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 23:37:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2762a58bad554ad193585a1bcb9b7c1a : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba292054bd75c0e62732bfc84baf099dfa313f19","revisions":{"ba292054bd75c0e62732bfc84baf099dfa313f19":{"kind":"REWORK","_number":1,"created":"2020-03-23 06:47:10.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/68/714368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/68/714368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/68/714368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/68/714368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/68/714368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/68/714368/1"}}},"commit":{"parents":[{"commit":"3d8df2147168b6a3820fd8f20e5cc6d8264968ed","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/3d8df2147168b6a3820fd8f20e5cc6d8264968ed"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-23 06:45:48.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-03-23 06:45:48.000000000","tz":480},"subject":"Remove idle_timeout option","message":"Remove idle_timeout option\n\nThe idle_timeout parameter has been deprecated for two releases.\nWe can remove it.\n\nChange-Id: I5beee515ab6ceb5df7ef54fd35189a5c72ba242f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/ba292054bd75c0e62732bfc84baf099dfa313f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/ba292054bd75c0e62732bfc84baf099dfa313f19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
