)]}'
{"id":"openstack%2Fpuppet-glare~715798","triplet_id":"openstack%2Fpuppet-glare~master~I3a21fbf091e3a3c9fc5f9b75edace2b67f9babc4","project":"openstack/puppet-glare","branch":"master","hashtags":[],"change_id":"I3a21fbf091e3a3c9fc5f9b75edace2b67f9babc4","subject":"Add CentOS 8 and Puppet 6 in metadata","status":"MERGED","created":"2020-03-30 07:22:54.000000000","updated":"2020-03-31 14:42:57.000000000","submitted":"2020-03-31 14:42:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715798-1585665777895-e8a3316c","meta_rev_id":"3639aa400de7dd14e0ad345d84b685db71e6dd51","_number":715798,"virtual_id_number":715798,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 14:42:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-30 13:53:14.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-31 14:31:10.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-31 14:31:10.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 13:53:14.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-31 14:31:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-31 14:42:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9713fd897f2d1d99eac5604a95ede090320521d2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-30 07:22:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"800a84acb98b6aa808613735e3138d6fb78d7d04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 07:31:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/32a10b90554441e7b954428b53f82716 : SUCCESS in 2m 18s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aea8c01015674cf6854c020aae3cca9f : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6ccebea161f5ebedbc2cc2f9798480ea9a540522","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-30 07:46:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"01346aebd7fccc5b38482fe1939e73f0c0c8748f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 08:01:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/b75a8771aea64d908282f81a06bf679a : SUCCESS in 2m 48s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf010d89ac22473e9a0f1c8c2de620b2 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"3c14d4da91d3719978ae93e0d8d234a57deba52a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-30 13:53:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bf14b8bf390991c1800e280e7dbab9f41dfe87c9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-31 14:31:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"09e14bfef7cf31da87c304376468653a1aa1cf32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:31:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a964249bc4644bd5b5f0762819ce14bebc297214","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:42:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/b9394954fe844699bce11224ca35de27 : SUCCESS in 2m 39s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9b78a3b7c644ce6aee1a13c8be89b1c : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":2},{"id":"0fb1519294e93f553fcc8d7e31f385782acec30e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:42:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d23bf5c3eea5f4c20fc0a53a9601fb674599cc2d","revisions":{"9cb97fcc2bf0fa95409dbcd61eed922638ea5c0b":{"kind":"REWORK","_number":1,"created":"2020-03-30 07:22:54.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/98/715798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/98/715798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/98/715798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/98/715798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/98/715798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/98/715798/1"}}},"commit":{"parents":[{"commit":"5ad72368352ac866a5bb00b8f2e8f907d1993dae","subject":"Deprecate min_pool_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/5ad72368352ac866a5bb00b8f2e8f907d1993dae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:22:46.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:22:46.000000000","tz":120},"subject":"Add CentOS 8 and Puppet 6 in metadata","message":"Add CentOS 8 and Puppet 6 in metadata\n\nChange-Id: I3a21fbf091e3a3c9fc5f9b75edace2b67f9babc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/9cb97fcc2bf0fa95409dbcd61eed922638ea5c0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/9cb97fcc2bf0fa95409dbcd61eed922638ea5c0b"}]},"branch":"refs/heads/master"},"d23bf5c3eea5f4c20fc0a53a9601fb674599cc2d":{"kind":"REWORK","_number":2,"created":"2020-03-30 07:46:51.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/98/715798/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/98/715798/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/98/715798/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/98/715798/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/98/715798/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/98/715798/2"}}},"commit":{"parents":[{"commit":"5ad72368352ac866a5bb00b8f2e8f907d1993dae","subject":"Deprecate min_pool_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/5ad72368352ac866a5bb00b8f2e8f907d1993dae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:22:46.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:46:49.000000000","tz":0},"subject":"Add CentOS 8 and Puppet 6 in metadata","message":"Add CentOS 8 and Puppet 6 in metadata\n\nChange-Id: I3a21fbf091e3a3c9fc5f9b75edace2b67f9babc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/d23bf5c3eea5f4c20fc0a53a9601fb674599cc2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/d23bf5c3eea5f4c20fc0a53a9601fb674599cc2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
