)]}'
{"id":"openstack%2Fpuppet-glare~726594","triplet_id":"openstack%2Fpuppet-glare~master~Ib80418c17151f06b7bf90171965163c736ba5dbc","project":"openstack/puppet-glare","branch":"master","topic":"nodepool-centos8","hashtags":[],"change_id":"Ib80418c17151f06b7bf90171965163c736ba5dbc","subject":"Add CentOS8 to nodesets","status":"MERGED","created":"2020-05-10 12:45:07.000000000","updated":"2020-08-26 12:18:51.000000000","submitted":"2020-05-11 22:27:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726594-1589236065126-6db76d6a","meta_rev_id":"1d63100c3ab4a17885afcd3d79a236dfba9b6b2f","_number":726594,"virtual_id_number":726594,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-11 22:27:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2020-05-11 07:44:32.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-05-11 22:14:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-05-11 22:14:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 07:44:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-11 22:14:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-11 22:27:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abfd29eb44f4e92aace5613d5866b718afe57e48","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-10 12:45:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86e91521dfaa81cad027799f787b07fff4ed8a28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 12:58:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a24aa7de94b841dfb557742230f40c55 : SUCCESS in 5m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c50512fde08401f916d37b40c26b429 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fdb948f872442cba7fd3c47e5ff8a8f : SUCCESS in 3m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1692b063daff4ab9b20ebe08636682fe : SUCCESS in 9m 22s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/08f42aa684e24ec5a2d45e48406ce88a : FAILURE in 7m 44s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ef544cca6cf49ff848685c0d13dc382 : FAILURE in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"576b5845df91a8012c39aaf4c9a08a9d12d9e549","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-11 07:44:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df419e3a0438257115f5740016c81c4a164ca15b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-11 22:14:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9038b3fff5400a493d47300b158f443886828ae7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 22:14:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b9fb9dfa4fc2d436be2e5de677d82f3b0551173","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 22:27:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ae94bccc4634e02904820fa9d7118cb : SUCCESS in 7m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ab87758a8254f769682c887530a8975 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fd7131a37fd450d929fe13dde336520 : SUCCESS in 5m 31s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2296c3b816e0454a9a20cea6c42100a2 : SUCCESS in 11m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6418b48541d5cb6732d2b5ee9f7045f266907cf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 22:27:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"59ff33e8279b080e4f5ce99ad3018a6e1fbce407","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-26 12:18:51.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 298dc99b5656f45ac5bfb36cdc6467e8318796bc","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2b9cdb704788476c6d692fba3d813b90afed162","revisions":{"a2b9cdb704788476c6d692fba3d813b90afed162":{"kind":"REWORK","_number":1,"created":"2020-05-10 12:45:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/726594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/94/726594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/94/726594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/94/726594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/94/726594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/94/726594/1"}}},"commit":{"parents":[{"commit":"147331b67428bf26217bf1d447888513acb52e86","subject":"Bump openstackdocstheme to 1.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/147331b67428bf26217bf1d447888513acb52e86"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-10 12:45:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-10 12:45:01.000000000","tz":540},"subject":"Add CentOS8 to nodesets","message":"Add CentOS8 to nodesets\n\nChange-Id: Ib80418c17151f06b7bf90171965163c736ba5dbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/a2b9cdb704788476c6d692fba3d813b90afed162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/a2b9cdb704788476c6d692fba3d813b90afed162"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
