)]}'
{"id":"openstack%2Fpuppet-glare~736884","triplet_id":"openstack%2Fpuppet-glare~master~I4d54789ab4404af28ccef09f51d04c2f6d34ec5d","project":"openstack/puppet-glare","branch":"master","topic":"prep-release-victoria-m1","hashtags":[],"change_id":"I4d54789ab4404af28ccef09f51d04c2f6d34ec5d","subject":"Prepare Victoria M1","status":"MERGED","created":"2020-06-19 02:40:37.000000000","updated":"2020-06-20 10:48:51.000000000","submitted":"2020-06-20 10:48:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736884-1592650131616-c15fc63c","meta_rev_id":"57091882a9db7ae0e856b3b647529d2c11813f3f","_number":736884,"virtual_id_number":736884,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-20 10:48:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-06-19 13:30:23.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-06-20 10:42:47.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-06-20 10:42:47.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 13:30:23.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-06-20 10:42:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-06-20 10:48:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54457dcdb276c904029383dbc1d9b3b570a399bb","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-06-19 02:40:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76bf59eccae15884938f0d3e4d10118be0e604bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 02:49:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/cbe3f5d92ba04f73930d2f3f130fceaa : SUCCESS in 3m 11s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a7f160edc7841fa87f1570be5c4ab21 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":1},{"id":"78c63593036ffb6d742b4db1fffe52941f529113","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-06-19 13:30:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9b865d247a9e4093cf7d6da550d6faa8b4c726b5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-06-20 10:42:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"278f1247220cc562224a5e67837a10ee87d5f2d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 10:42:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1395d675d5e620ace091ec47ff25273fffef13f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 10:48:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d889d33ec03040a188717228b78fdfa83ce41d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 10:48:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/f28cebb522834282b8fbef63e9342ec5 : SUCCESS in 3m 07s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66635e185c8e4c738d1e41843deb6231 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"116c4a5fb49a135ca61ea7ffd9a4ea82ee16fa81","revisions":{"116c4a5fb49a135ca61ea7ffd9a4ea82ee16fa81":{"kind":"REWORK","_number":1,"created":"2020-06-19 02:40:37.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/84/736884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/84/736884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/84/736884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/84/736884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/84/736884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/84/736884/1"}}},"commit":{"parents":[{"commit":"9a9b3a4e0bb8a9bba21fb03fae21a50ba504f9a3","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/9a9b3a4e0bb8a9bba21fb03fae21a50ba504f9a3"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-06-19 02:38:54.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-06-19 02:38:54.000000000","tz":480},"subject":"Prepare Victoria M1","message":"Prepare Victoria M1\n\nUpdate the version metadata for Victoria milestone M1\n\nChange-Id: I4d54789ab4404af28ccef09f51d04c2f6d34ec5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/116c4a5fb49a135ca61ea7ffd9a4ea82ee16fa81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/116c4a5fb49a135ca61ea7ffd9a4ea82ee16fa81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
