)]}'
{"id":"openstack%2Fpuppet-glare~741353","triplet_id":"openstack%2Fpuppet-glare~master~Idbaf877352d93a9713a87b4ceadd18feb770ad6d","project":"openstack/puppet-glare","branch":"master","topic":"at-interface","hashtags":[],"change_id":"Idbaf877352d93a9713a87b4ceadd18feb770ad6d","subject":"Add support for the interface parameter in authtoken middleware","status":"MERGED","created":"2020-07-16 01:27:43.000000000","updated":"2020-07-17 00:01:47.000000000","submitted":"2020-07-17 00:00:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741353-1594944027202-56289871","meta_rev_id":"d0d209a217a8077881387c63552396dca563f617","_number":741353,"virtual_id_number":741353,"owner":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-17 00:00:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-07-16 14:14:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-07-16 02:53:16.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-07-16 14:14:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 02:53:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-07-16 14:14:11.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-07-17 00:00:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21b55f2f384addfbe2c6f2833913f6fbf6fbb368","author":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"date":"2020-07-16 01:27:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f823dea0b9e0b67ef2f479316ddbc12ece1e6b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 01:44:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22c1f7b483d248a1a9752dfcf84ea35f : SUCCESS in 5m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31757535847d42a6857d17747168f27c : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/872a2d2c80bf420b8d082a1e8c1f7656 : SUCCESS in 4m 19s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/135aaac5f53c4e2d8090accfa2087cb5 : SUCCESS in 10m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c43f8df6e6704e9e99965b2d1c4b3e20 : SUCCESS in 10m 52s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f37b75fcef4c4fbf9416caf4d260f23d : SUCCESS in 8m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb13548942c54b5199bc5b2b170af3b3 : FAILURE in 7m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2437b34ef6a648e1ac7da96c9432ccf1 : SUCCESS in 11m 19s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0b020c6ce79e4be3aff95e39a9f0b54d : FAILURE in 8m 30s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4439b5a98b8747b7ac0807d4db5b2a01 : FAILURE in 8m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d7a2dfcd1f543f7bd0289a2ab6281e6 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"066c38885750d475ec401cca0f69c87d39735157","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-16 02:53:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b87f0cba7150ff8a780f52397573f6a7f11b43ad","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-16 14:14:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c239c1a24d1f9e874450f29d40ea53161093436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 14:14:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"306e5b041c0dd7683639fc5a3fb3071ec50d9e97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 14:38:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ff656a1c7bc4040b035632ed8832de4 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/141e87088ec1416b95c0eda1560fdec5 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7cdd6150cda47b28a1abce0c2dc0279 : SUCCESS in 5m 09s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/386ab89ec6e74212a0fc5cc8f8360573 : SUCCESS in 9m 57s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b4af02b14736426c9194c5e0779d2c80 : SUCCESS in 13m 21s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bcb8ee622384f90b64e8a54ec6fa9df : SUCCESS in 9m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/18e0e5572e15446d80c620b6a8d3941f : FAILURE in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44f0e61d08bd4ab490a8aca8fb5fafec : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"bc1a18c0b05bc3bcf30b90a1221a08a54a0d9a53","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-16 23:27:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7a8a2fc4e36a78a0823b4dc26e28d25c85f54dc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 23:44:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f7dcba9b0eb48ad88b49abd3069d490 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5f1dfe9248f4ca1b351dca0b1199158 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8043c372dbe8446b9cb525cf871b9a97 : SUCCESS in 4m 22s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/852b0f0e0d1243428bff0f5cbcd6111b : SUCCESS in 10m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ac87438d21d3481aa7a8655412dce6c7 : SUCCESS in 10m 52s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc69ed0857c34399b2cbc1bba34b1e4c : SUCCESS in 9m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2b671ec577a4964959398249d444b16 : FAILURE in 7m 35s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/30351d4c2c494693b349babaf9981073 : SUCCESS in 10m 02s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/fb484012a6d14ec6af461333cea60f9d : FAILURE in 8m 09s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/781e9fd4fd194c20ae4bf2d6434047cf : FAILURE in 6m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/205d577647134dc783dd2d4727548beb : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"27f5e55509f24ed5712c572efd47404be5b86295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 23:44:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c4b1217cdd229ce4fdde3ca9405670a0df0ffa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 00:00:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fd799f0dec84b399a41610255230904 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f0b2e826f604f60b9df9f54365f06e1 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a77d2caca8f4e53a937cc9e0d54c949 : SUCCESS in 4m 16s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/fae5383eca504fbd8c79934095c16f5d : SUCCESS in 10m 01s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/71760d12b0c5436ebc9352573fcb3e8c : SUCCESS in 10m 00s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7979d57c06f747d79b382c993246dd08 : SUCCESS in 8m 16s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/380d7d69fa4743d0817e13ece63e7fea : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7e7b3130340483e89495eb89a34ddc3 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0108b4b20ca7e7ea600a79df348310af8263815f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 00:00:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4970601ca12e1e88b73c8a37ddc4cc08b82fa71b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 00:01:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fe13fc13c9c4407be5b8ba955bc8342 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ede6fbf786fce99cf2f57c58ad5923fd4daa7593","revisions":{"ede6fbf786fce99cf2f57c58ad5923fd4daa7593":{"kind":"REWORK","_number":1,"created":"2020-07-16 01:27:43.000000000","uploader":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"ref":"refs/changes/53/741353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/53/741353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/53/741353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/53/741353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/53/741353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/53/741353/1"}}},"commit":{"parents":[{"commit":"116c4a5fb49a135ca61ea7ffd9a4ea82ee16fa81","subject":"Prepare Victoria M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/116c4a5fb49a135ca61ea7ffd9a4ea82ee16fa81"}]}],"author":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-16 01:26:20.000000000","tz":600},"committer":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-16 01:27:40.000000000","tz":600},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: Idbaf877352d93a9713a87b4ceadd18feb770ad6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/ede6fbf786fce99cf2f57c58ad5923fd4daa7593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/ede6fbf786fce99cf2f57c58ad5923fd4daa7593"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
