)]}'
{"id":"openstack%2Fpuppet-glare~747089","triplet_id":"openstack%2Fpuppet-glare~master~I29e8d699eb34c4d89302d9b0ff08562353c1e4a1","project":"openstack/puppet-glare","branch":"master","topic":"bug/1892284","hashtags":[],"change_id":"I29e8d699eb34c4d89302d9b0ff08562353c1e4a1","subject":"Add service_token_roles for keystone authtoken config","status":"MERGED","created":"2020-08-20 02:42:47.000000000","updated":"2020-08-20 21:42:31.000000000","submitted":"2020-08-20 21:39:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747089-1597959599110-0b925b31","meta_rev_id":"ac0cf69621790994f5a3dba17e887f32e58afa4c","_number":747089,"virtual_id_number":747089,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2020-08-20 20:51:34.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-20 21:39:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-20 12:11:07.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-20 14:33:51.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-20 14:33:51.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 12:11:07.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-20 14:33:51.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-08-20 20:51:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-20 21:39:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c01d9433e2a2bb53dfbef1b4acd8fa2f29b98eec","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-20 02:42:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1ccef99cf53e2bd3f18518b0fa5dafb692cddb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 03:07:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5baecf53d0c74840913eb8ee82e6e4e7 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b93f9f7b76b44b9d878436c12b04d89d : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5113b8d30064475190099df9adbf63f2 : SUCCESS in 4m 32s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3a20957021a94b4e82c51dcb1563feeb : SUCCESS in 9m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9349f484fe684843856307ae40e65bed : SUCCESS in 15m 01s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd6f0f1c30a0416ba2ba84465f9d1996 : SUCCESS in 12m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49de12f6859a4ddc97dc8ab4caa2b658 : FAILURE in 8m 29s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fb7a1b68013c46b6ac1cc4077448b2ca : SUCCESS in 11m 36s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/cac3a19bb05e457bb8be6745297e5140 : FAILURE in 9m 19s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7373a73f9dc249278919cd5464732f75 : FAILURE in 7m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff7132d15f0847e8a4d6d8bdb98d241c : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e378c31fcb48b57e2b3037e6329d14803a391020","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-20 12:11:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1f743f78052043f78b3893b18ab98167ab965de","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-20 14:33:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c594c607a74cedeb3ee8ce075232d9e4fab135fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 14:36:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"699a584102030defe2be5db401445f2c93fc73af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 15:15:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7eb34526ac56468ea33b7f45dcf7d499 : RETRY_LIMIT in 1m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99a572f734224093a9fb6309daee1fa2 : RETRY_LIMIT in 1m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/856bad545de848d3a6af206cb7d59acd : RETRY_LIMIT in 2m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/0c889ca3dc3f4187a017055e30511150 : SUCCESS in 14m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4d7215811ed14174b6d9c0089e1b17cc : SUCCESS in 11m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f32d5ec89e7648a18530097047d9c663 : RETRY_LIMIT in 3m 55s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fdc8c7dc1ac1425297374e4fbaec2485 : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65f1583b21a7479588291cb9e2e5d176 : RETRY_LIMIT in 1m 32s","accounts_in_message":[],"_revision_number":1},{"id":"de4e2ac7104206eff1e52118b2c8266ec78f4a73","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-20 20:51:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"275ff0cfccf8ec61cf9add16b6ac13c65755765b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:17:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ab1416060104844b94c66256d5fdb1b : SUCCESS in 4m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7d80dbaf78447138a128ef3cd0252e6 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5f962719e26454d8117af315168c31f : SUCCESS in 5m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ac859b6de5374b9cb490551a8444a01d : SUCCESS in 11m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3bd1815fe6f4440eaa23ee9b5c47a0d6 : SUCCESS in 10m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d83caa4675204247b6dc2a25ade014eb : SUCCESS in 12m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/375917d3168f4e28856de8067d74718a : FAILURE in 7m 35s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8a8d209254cc400493087511c2cd9532 : SUCCESS in 12m 09s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/640eefb6dce14db1a8c4c19f834c046f : FAILURE in 8m 12s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee776a934e53448b9a7acaf9dd622ba9 : FAILURE in 9m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e92d24360644c67a825ed0fa32a8985 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7ab600eed2e1dbf8f2ee5e90dfdb6d48ae97eace","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:17:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98cfbc587dfe52963b653fb8eac1ba5d706c9c90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:39:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88287fb70bc14561840543c7d56c0fdf : SUCCESS in 5m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20bd0428d1ab4b6ab15bdee23f55270d : SUCCESS in 4m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76a05f222a9e4974a7c2da46457b9ad6 : SUCCESS in 4m 35s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/956b3dfeaf3142cb833bab7b4d9dca2d : SUCCESS in 10m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/411a28549cbf4fd7bcd21c07dfd062c0 : SUCCESS in 11m 54s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0ca5e9da7a6440684e407399742bf32 : SUCCESS in 10m 02s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3f649e6c4efe4890bc27e9201c7efcb4 : SUCCESS in 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2cb17efa3794921955b9d54dba31a5b : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f07f04e1533a77b37ab83b86a966e4a06975337d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:39:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"85db674f2106a11d001262343de066ac85677a29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:42:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cac555a6ca6045988f5b250da63aa2c8 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6034694981001c35595a47940954cb160563fa6e","revisions":{"6034694981001c35595a47940954cb160563fa6e":{"kind":"REWORK","_number":1,"created":"2020-08-20 02:42:47.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/89/747089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/89/747089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/747089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/747089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/747089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/89/747089/1"}}},"commit":{"parents":[{"commit":"38989934bf49e3e6c92108608fd65d5f762de763","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/38989934bf49e3e6c92108608fd65d5f762de763"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-20 02:41:27.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-20 02:41:27.000000000","tz":480},"subject":"Add service_token_roles for keystone authtoken config","message":"Add service_token_roles for keystone authtoken config\n\nAdd the ability to configure service_token_roles.\n\nChange-Id: I29e8d699eb34c4d89302d9b0ff08562353c1e4a1\nCloses-Bug: #1892284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/6034694981001c35595a47940954cb160563fa6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/6034694981001c35595a47940954cb160563fa6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
