)]}'
{"id":"openstack%2Fpuppet-glare~748893","triplet_id":"openstack%2Fpuppet-glare~master~I47d4a6845ed394e6666d9f9db6796d473514e14b","project":"openstack/puppet-glare","branch":"master","hashtags":[],"change_id":"I47d4a6845ed394e6666d9f9db6796d473514e14b","subject":"Use openstack_spec_helper from zuul checkout","status":"MERGED","created":"2020-08-30 20:56:23.000000000","updated":"2020-08-31 13:56:14.000000000","submitted":"2020-08-31 13:56:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748893-1598882174002-8a9bf4a0","meta_rev_id":"a997b0401bf23a3f0e7b8df921e57d10bfcfffdc","_number":748893,"virtual_id_number":748893,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-31 13:56:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-31 13:30:15.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-31 00:53:55.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-31 13:30:15.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 00:53:55.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-08-31 13:30:15.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-31 13:56:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"306b489de74e4602093fb0ef3dfeb3d12f51a0e3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-30 20:56:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ccd83b6b9281f5f2650eb2d5d5e2a05a500fd37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-30 21:12:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/866605eafdcb47aa90cef0fb48a1ebd9 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5491f99ff67426c9c7dc5d83e8b3081 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/238863722ddf4765b53939333fcdc2e6 : SUCCESS in 2m 24s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5244383c3128466c910efe72c704d2da : SUCCESS in 8m 16s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/927eade9d398457482e8f0b0e62f179c : SUCCESS in 8m 57s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a784679a90d4445832d81247305e535 : SUCCESS in 6m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f612ef496d9e4b56a58f0030acbbe64c : SUCCESS in 6m 43s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ba8f7d3bd2514a75bc1b796d7585c9b1 : SUCCESS in 7m 39s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a10434e3ea304cca9a9fc2b03b7a0683 : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cfc3e1d75115d0b979a1cb688bc70deb9abfeeb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 00:53:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"be4cd8c39c536c1fceb384c8d943f255cbb59ffa","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-31 13:30:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"384d9b707b43f68f3488fd4c60b513fcceba7de3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:30:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"509c6509ee64dd844658c2b5fc1fc31ace847040","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:56:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2c1e14ddecb483a818698e06370259b : SUCCESS in 4m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/317aed9cbca94c118ce36a740c028c59 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e162d30415442899bd944005af524a7 : SUCCESS in 3m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/714c3001832247339ce619d40c859ea2 : SUCCESS in 9m 40s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2054d159137845f287afba4c3a05e639 : SUCCESS in 12m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65634225ab694b0a88424e1f1c650ecd : SUCCESS in 6m 47s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/a35b48a442f3437f9185ccc2f6d08abe : SUCCESS in 9m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a997b0401bf23a3f0e7b8df921e57d10bfcfffdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:56:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fbaf7651395d1cbdd098dc702c3c3d06fc5d321","revisions":{"4fbaf7651395d1cbdd098dc702c3c3d06fc5d321":{"kind":"REWORK","_number":1,"created":"2020-08-30 20:56:23.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/93/748893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/93/748893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/93/748893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/93/748893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/93/748893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/93/748893/1"}}},"commit":{"parents":[{"commit":"6dfb89f4cef59c4b1a390bcd789f351affc9a692","subject":"Add mysql_enable_ndb option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/6dfb89f4cef59c4b1a390bcd789f351affc9a692"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-30 20:56:16.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-30 20:56:16.000000000","tz":120},"subject":"Use openstack_spec_helper from zuul checkout","message":"Use openstack_spec_helper from zuul checkout\n\nChange-Id: I47d4a6845ed394e6666d9f9db6796d473514e14b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/4fbaf7651395d1cbdd098dc702c3c3d06fc5d321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/4fbaf7651395d1cbdd098dc702c3c3d06fc5d321"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
