)]}'
{"id":"openstack%2Fpuppet-glare~760009","triplet_id":"openstack%2Fpuppet-glare~master~I2a7d787dbbd0aeef8f2730b1a75235f4f06b566f","project":"openstack/puppet-glare","branch":"master","topic":"fix-source-metadata","hashtags":[],"change_id":"I2a7d787dbbd0aeef8f2730b1a75235f4f06b566f","subject":"Fix outdated files","status":"MERGED","created":"2020-10-28 03:03:23.000000000","updated":"2020-10-28 03:41:45.000000000","submitted":"2020-10-28 03:41:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760009-1603856505082-ea02ee25","meta_rev_id":"e7786cf262d610ef51f356bf7cf79da29fcdd6fc","_number":760009,"virtual_id_number":760009,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-28 03:41:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-28 03:33:18.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-10-28 03:33:18.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 03:41:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c7c960d574a8c44d09b237a1342ec391e2943ae","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-28 03:03:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce34adb32222c0cf0940a135ac8c2581249fea4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 03:14:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/ee889782b3934fbeba4d2fce0aa8541e : SUCCESS in 2m 16s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5bececdd0b14b3fb830eb4e6c8c1121 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8c24114cc3268be3a700f6372d700130928dc838","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-28 03:33:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"758ba6e70c19f95d0d50474aeed474d354adfdf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 03:33:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f08607c37975d8ddf620c2fbaf194f98195fc246","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 03:41:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/635777861438447a968515ca5de52c27 : SUCCESS in 3m 41s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad285e99a1e74643b617afceda07d5f7 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"21daeee1503e1aa64cc1b0371a92be7b06c3edf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 03:41:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"654bafe16dcb8f1f9674af181038cd49c6f9105c","revisions":{"654bafe16dcb8f1f9674af181038cd49c6f9105c":{"kind":"REWORK","_number":1,"created":"2020-10-28 03:03:23.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/09/760009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/09/760009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/09/760009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/09/760009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/09/760009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/09/760009/1"}}},"commit":{"parents":[{"commit":"ad0d02353d80ca9807ec8ebbcdca8affbf5d4031","subject":"Bump version for the start of Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/ad0d02353d80ca9807ec8ebbcdca8affbf5d4031"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-28 02:40:09.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-28 02:40:09.000000000","tz":480},"subject":"Fix outdated files","message":"Fix outdated files\n\nChange-Id: I2a7d787dbbd0aeef8f2730b1a75235f4f06b566f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/654bafe16dcb8f1f9674af181038cd49c6f9105c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/654bafe16dcb8f1f9674af181038cd49c6f9105c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
