)]}'
{"id":"openstack%2Fpuppet-glare~761089","triplet_id":"openstack%2Fpuppet-glare~master~Iacf60927f83761973b77dd9b5f876b14a58eb0d6","project":"openstack/puppet-glare","branch":"master","topic":"at-params","hashtags":[],"change_id":"Iacf60927f83761973b77dd9b5f876b14a58eb0d6","subject":"Add support for the keystone_authtoken/service_type parameter","status":"MERGED","created":"2020-11-03 08:54:04.000000000","updated":"2020-11-07 22:23:13.000000000","submitted":"2020-11-07 22:22:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761089-1604787722981-b2dd4156","meta_rev_id":"2067c851c5fbef53d2adb8841f9141088543fdc5","_number":761089,"virtual_id_number":761089,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-07 22:22:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2020-11-03 09:43:35.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-11-07 22:06:11.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-11-07 22:06:11.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-03 09:43:35.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-11-07 22:06:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-07 22:22:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29f9ff029059b7dfc09eb8e52fe0c46579bc2482","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-03 08:54:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49f52a3b2be53b5e1916a26ae4ce5beb9269e645","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:27:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c26201aaa5a346f3ad5f2f11333db3df : SUCCESS in 2m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f28f9715710b4437aca060b431eff671 : SUCCESS in 3m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/34e2fec163b34eb09d9299c872648652 : SUCCESS in 9m 37s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a2e5f335ba14d2297ecc07f5708a571 : SUCCESS in 11m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdc5e945d2384ebb95459bc0795f6c89 : SUCCESS in 7m 17s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/5b773b4dc72b43918ca870fb0fcac121 : SUCCESS in 8m 33s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbafc92f292747b3a4b1c4befb5b3093 : SUCCESS in 5m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5a80436ea1b46ee94850596283c55e9 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"832b185133b5f6d355ce9728908aea7135d34ec7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-11-03 09:43:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"302258e0effa7c3117cfaf25acf0c57d55ae7574","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-07 22:06:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f633fcd22ca1080fc5b6ec27b444f1716e212b15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 22:06:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a71438ed4fb2fb3da2b94e27526dbd6091c87ee7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 22:22:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b3b1c8c0cdb9537c7c00b1770192372cdff64f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 22:22:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/582c9e16410f45b3b92f1f8d49567517 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b4a3c99e5854c46b039211ce142200c : SUCCESS in 2m 38s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8ce7458fe7e345e6b2afc7beb1c1a6d5 : SUCCESS in 8m 13s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/236cff76e8a947faa686cd78bf2ec5b0 : SUCCESS in 5m 49s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/7220b2f0b4b84b15833b0ff54cca756a : SUCCESS in 7m 58s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e22fcec7644a45e9b3044cf8e1b33891 : SUCCESS in 5m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a78c9942f1c4adaa4cac8474da751a1 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"70e56177cc02662c7fcfdcc33c2d1f79eaf733d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 22:23:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7adfdf202ad242c2a47feccd9ab9121e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba039b45ecfe9bcd15fd8a73f1df89024e32a6c1","revisions":{"ba039b45ecfe9bcd15fd8a73f1df89024e32a6c1":{"kind":"REWORK","_number":1,"created":"2020-11-03 08:54:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/761089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/89/761089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/761089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/761089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/89/761089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/89/761089/1"}}},"commit":{"parents":[{"commit":"ad0d02353d80ca9807ec8ebbcdca8affbf5d4031","subject":"Bump version for the start of Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/ad0d02353d80ca9807ec8ebbcdca8affbf5d4031"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-03 08:53:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-03 08:53:49.000000000","tz":540},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: Iacf60927f83761973b77dd9b5f876b14a58eb0d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/ba039b45ecfe9bcd15fd8a73f1df89024e32a6c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/ba039b45ecfe9bcd15fd8a73f1df89024e32a6c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
