)]}'
{"id":"openstack%2Fpuppet-glare~765257","triplet_id":"openstack%2Fpuppet-glare~stable%2Fussuri~Idbaf877352d93a9713a87b4ceadd18feb770ad6d","project":"openstack/puppet-glare","branch":"stable/ussuri","topic":"at-interface-stable/ussuri","hashtags":[],"change_id":"Idbaf877352d93a9713a87b4ceadd18feb770ad6d","subject":"Add support for the interface parameter in authtoken middleware","status":"MERGED","created":"2020-12-03 06:34:53.000000000","updated":"2020-12-04 07:11:52.000000000","submitted":"2020-12-04 07:10:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765257-at-interface-stable/ussuri","cherry_pick_of_change":741353,"cherry_pick_of_patch_set":1,"meta_rev_id":"aeb7f287397598c9bb013898773b19e2ab3d476b","_number":765257,"virtual_id_number":765257,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 07:10:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-12-04 06:39:53.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-12-03 21:58:12.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-12-04 06:39:53.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"}],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-03 06:34:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"state":"REVIEWER"},{"updated":"2020-12-03 08:27:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 21:58:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-12-04 06:39:53.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"b5cd563bf154e97a175cd2b2ad06d92209332794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-03 06:34:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6f6ae57b55df69ae9a89e49b95a1e248cbb99972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-03 06:48:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a158d1cca09b230f60de0c7045b904d047b1c541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 08:27:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aaa0f899b6549948e3d333ad3d15a8f : SUCCESS in 3m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d38365f65e3946b89e79dc2cf9e49a3a : SUCCESS in 2m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd70a29a57874fd58fbdcc9c77ffe09a : SUCCESS in 2m 57s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/35ed0b52320f44c2b57001d0a836eae5 : SUCCESS in 19m 32s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/3f5bc5e4617f4a49983a3da7194c6b7d : SUCCESS in 12m 56s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ffe4e9a4ee81415e93683757be778afe : SUCCESS in 9m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94588697437d4325b444daebe5b5713d : SUCCESS in 7m 31s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/f18a56dffee946658e3a6099ac0055c8 : SUCCESS in 9m 21s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a607fab3813043b7800e1e4b7d72f6a6 : SUCCESS in 6m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3edf3a7fd1740078a7ae2087a7e3eb4 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"e581fe35a1c46016dfedc27d6aea29152e5838bb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-12-03 21:58:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e219a602d6308be3610d75b9dbfc3abaebd28357","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-12-04 06:39:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7d95a73194712d634bc21c142c98898ae346100f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 06:40:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"152c80803f6cdf390b1c6b16da0751690a096fa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 07:10:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9af47abbd3984005a7cffacef1ce42ce : SUCCESS in 2m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4ca1fc8ded345ee8d446273c27e5810 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82caa95ef87c49a0b893905304965c8f : SUCCESS in 7m 43s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3bba06524fac4fb39626597f1d6158b6 : SUCCESS in 10m 36s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/8b3d5c86014b4f6e93b493068d9250c4 : SUCCESS in 8m 45s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2ae69d849f24d09b74d9c85c11d7982 : SUCCESS in 7m 07s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/5c809b7568e8445b95431477a4606879 : SUCCESS in 12m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98ca4b4134f243b19b261204213f9324 : SUCCESS in 9m 15s","accounts_in_message":[],"_revision_number":2},{"id":"fd6bb68771d2195202abcc065562a0cb1b8a62f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 07:10:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"aeb7f287397598c9bb013898773b19e2ab3d476b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 07:11:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95874dfe70704df5967adb5ae65b5604 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"083026ac23e8368ec95fe6678f8aa1056f2e47e9","revisions":{"831c4626b3ae87e3f6d11107281eba96eef4b66d":{"kind":"REWORK","_number":1,"created":"2020-12-03 06:34:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/765257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/57/765257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/57/765257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/57/765257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/57/765257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/57/765257/1"}}},"commit":{"parents":[{"commit":"298dc99b5656f45ac5bfb36cdc6467e8318796bc","subject":"Add CentOS8 to nodesets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/298dc99b5656f45ac5bfb36cdc6467e8318796bc"}]}],"author":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-16 01:26:20.000000000","tz":600},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-03 06:34:53.000000000","tz":0},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: Idbaf877352d93a9713a87b4ceadd18feb770ad6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/831c4626b3ae87e3f6d11107281eba96eef4b66d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/831c4626b3ae87e3f6d11107281eba96eef4b66d"}]},"branch":"refs/heads/stable/ussuri"},"083026ac23e8368ec95fe6678f8aa1056f2e47e9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-03 06:48:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/765257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/57/765257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/57/765257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/57/765257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/57/765257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/57/765257/2"}}},"commit":{"parents":[{"commit":"298dc99b5656f45ac5bfb36cdc6467e8318796bc","subject":"Add CentOS8 to nodesets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/298dc99b5656f45ac5bfb36cdc6467e8318796bc"}]}],"author":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-16 01:26:20.000000000","tz":600},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-03 06:48:20.000000000","tz":0},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: Idbaf877352d93a9713a87b4ceadd18feb770ad6d\n(cherry picked from commit ede6fbf786fce99cf2f57c58ad5923fd4daa7593)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/083026ac23e8368ec95fe6678f8aa1056f2e47e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/083026ac23e8368ec95fe6678f8aa1056f2e47e9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
