)]}'
{"id":"openstack%2Fpuppet-glare~765675","triplet_id":"openstack%2Fpuppet-glare~master~I0c44a831aac880b7be631c573dfb76aee116c614","project":"openstack/puppet-glare","branch":"master","topic":"prep-release-wallaby-m1","hashtags":[],"change_id":"I0c44a831aac880b7be631c573dfb76aee116c614","subject":"Prepare Wallaby M1","status":"MERGED","created":"2020-12-06 03:04:39.000000000","updated":"2020-12-12 03:44:13.000000000","submitted":"2020-12-12 03:44:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765675-prep-release-wallaby-m1","meta_rev_id":"53f36c7e34fa85d697bea0582399c4d45d46d9bb","_number":765675,"virtual_id_number":765675,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-12 03:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-12-12 03:38:47.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-12-08 15:50:38.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-12-12 03:38:47.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-06 03:18:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 15:50:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a47f5625d8edb502dc7fd4afb7645cc87f2a9a2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-12-06 03:04:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a64ecb474d6eb35f62e8d63ad1ac7a18b1f3d1b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 03:18:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/b4fb84904eda4d6abe6751e4f4456d27 : SUCCESS in 2m 10s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3436fb61510d43108793631974b4b5be : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4357d3e5ec23fd6f369b42b985da222bc95a59e2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-12-08 15:50:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"39fa279599a87dcc4967c2f316bc0d8c7ea9cf8a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-12-12 03:38:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d013ff4c313ee9c068e239299ff48282a266d55e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 03:38:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b622ec349adf4f146a3745046aacf5676667e9cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 03:44:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/de394b9c78834e6680ad0237a4cfedd5 : SUCCESS in 2m 52s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d7404af4d9248059f6a4cf0f509e90b : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"53f36c7e34fa85d697bea0582399c4d45d46d9bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 03:44:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6a2634f04b0a9eb7320512e7526363d96a1d2d7","revisions":{"a6a2634f04b0a9eb7320512e7526363d96a1d2d7":{"kind":"REWORK","_number":1,"created":"2020-12-06 03:04:39.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/75/765675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/75/765675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/75/765675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/75/765675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/75/765675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/75/765675/1"}}},"commit":{"parents":[{"commit":"d33230e6bece69ed3b77335375599a5baedf71ed","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/d33230e6bece69ed3b77335375599a5baedf71ed"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-12-06 03:04:00.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-12-06 03:04:00.000000000","tz":480},"subject":"Prepare Wallaby M1","message":"Prepare Wallaby M1\n\nUpdate the version metadata for Wallaby milestone M1\n\nChange-Id: I0c44a831aac880b7be631c573dfb76aee116c614\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/a6a2634f04b0a9eb7320512e7526363d96a1d2d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/a6a2634f04b0a9eb7320512e7526363d96a1d2d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
