)]}'
{"id":"openstack%2Fpuppet-glare~782647","triplet_id":"openstack%2Fpuppet-glare~master~I522ac2fb7fe289cbcda436119a003850efce2aeb","project":"openstack/puppet-glare","branch":"master","topic":"enforce_new_defaults","hashtags":[],"change_id":"I522ac2fb7fe289cbcda436119a003850efce2aeb","subject":"Add support for oslo_policy/enforce_new_defaults","status":"MERGED","created":"2021-03-24 07:36:20.000000000","updated":"2021-03-25 01:44:04.000000000","submitted":"2021-03-25 01:42:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782647-enforce_new_defaults","meta_rev_id":"4a69f0732873b2abef1bae7f97cb1510a37a55ed","_number":782647,"virtual_id_number":782647,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 01:42:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-03-25 01:26:33.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-03-24 14:48:06.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-03-25 01:26:33.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 07:52:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-24 14:48:06.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-03-25 01:26:33.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"78fac38659b435bb79e28da1a08ac65e42b2500e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-24 07:36:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f596b2ee0d78e4849cb7655ac8cb1e7cdb71b52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 07:52:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e789659eeb44d2d9c92f1ffc90ff93a : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50d19a70e8894c55822bab5fbf81d117 : SUCCESS in 3m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/37c722d84bd141f7b65badd0b275337d : SUCCESS in 9m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7d0101c6d4c4bfc932283efb519235b : SUCCESS in 5m 51s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae60a2515b2e48528587ada2f38a44ed : SUCCESS in 5m 51s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e07cb41f2a144dd89dc4310fa00b2e5 : SUCCESS in 8m 24s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c218a59b2ebd4be9bf5b584234b53db3 : SUCCESS in 6m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20497f4422db4b3585d73eafd27aa853 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"efa50a37d38c5fcb67f26b22f41a9120247ba754","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-24 14:48:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e32361ba1b4f73fe75e7ed058fde6518e856af7a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-25 01:26:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3ade9dfdd5ecd04c6e7594708d09d18cc8f0a09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 01:27:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f306f04022f27e610e629b38cecd131afb84d24f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 01:42:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78611795e2d2487dbe92041254410a9c : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b6399b9cd0e44159c201fdccb50ba0a : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6dc4eedeb79346f2bd9d6798a4633fab : SUCCESS in 9m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08006756cbd14c1badd90cd55bc75e30 : SUCCESS in 6m 24s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/2207c3505c7c44efb08eda072ddcfe8d : SUCCESS in 8m 01s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69f0b36dec4748b791cbc448002524f5 : SUCCESS in 8m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07d85b954f694d188739ea802670d312 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"858cc531c94426bef6f08a9fe5141bbd6c06d3d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 01:42:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4a69f0732873b2abef1bae7f97cb1510a37a55ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 01:44:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/418395cb6d95453aa997b6808bae1dd3 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a41b6ee883c75103c35c920dbb779b3f7b180f2","revisions":{"2a41b6ee883c75103c35c920dbb779b3f7b180f2":{"kind":"REWORK","_number":1,"created":"2021-03-24 07:36:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/782647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/47/782647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/47/782647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/47/782647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/47/782647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/47/782647/1"}}},"commit":{"parents":[{"commit":"18c70a3a69dc8ec24e0a415c89c08cacb3ad053b","subject":"Add support for the oslo_policy/enforce_scope parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/18c70a3a69dc8ec24e0a415c89c08cacb3ad053b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 07:36:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 07:36:09.000000000","tz":540},"subject":"Add support for oslo_policy/enforce_new_defaults","message":"Add support for oslo_policy/enforce_new_defaults\n\nDepends-on: https://review.opendev.org/781428\nChange-Id: I522ac2fb7fe289cbcda436119a003850efce2aeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/2a41b6ee883c75103c35c920dbb779b3f7b180f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/2a41b6ee883c75103c35c920dbb779b3f7b180f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
