)]}'
{"id":"openstack%2Fpuppet-glare~790081","triplet_id":"openstack%2Fpuppet-glare~master~Id76e00fafd171f3d77b70817c5cfd103cb207b0e","project":"openstack/puppet-glare","branch":"master","topic":"retire-puppet-glare","hashtags":[],"change_id":"Id76e00fafd171f3d77b70817c5cfd103cb207b0e","subject":"Retire puppet-glare - Step 2: Remove Project Content","status":"MERGED","created":"2021-05-06 14:55:52.000000000","updated":"2021-05-11 15:11:23.000000000","submitted":"2021-05-11 15:11:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790081-retire-puppet-glare","meta_rev_id":"a34be2bdb588d59155bf50eea029034a55ee51fa","_number":790081,"virtual_id_number":790081,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-11 15:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2021-05-11 15:09:58.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-10 19:51:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2021-05-11 15:09:58.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 15:20:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 15:51:24.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-05-10 19:51:53.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"8b142c37c6a9c6028dafdefb15976a3b1682361c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-06 14:55:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6394b3919e54a006c7419beed444a9f874d040f2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-06 14:56:22.000000000","message":"Topic retire-glare removed","accounts_in_message":[],"_revision_number":1},{"id":"00fa8a937f6a78cf43bce1aca75ce5a6da8a3859","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-06 14:56:28.000000000","message":"Topic set to retire-puppet-glare","accounts_in_message":[],"_revision_number":1},{"id":"b36a5e35d181a849a7a5dfa1a96b6ea3473dec0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:20:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/48096028c2224b3f89c0691b3bec2366 : FAILURE in 3m 55s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/873c610e112549e8bdf2bbe8d692b163 : FAILURE in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8548d693c0440faadcbf69784919561 : FAILURE in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c15b8ab9b804225865ef7fae3b8fe37 : FAILURE in 4m 34s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/30d9482a4f71403a9a85a7a3801c7ef3 : FAILURE in 9m 29s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/522ecaf9158347e6af80d7b6203dfbbf : FAILURE in 9m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69501dd8624b490daf874cefa7afd68e : FAILURE in 7m 46s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/904be52a85f14cad8f0ec0719262fc6b : FAILURE in 6m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5b50dfa0a3947168c631e271db755a3 : FAILURE in 9m 10s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b566f13932f54c7b94f14200eaed9ec5 : FAILURE in 8m 40s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/155d9f24fff743bb8689b329d69eee1d : FAILURE in 5m 50s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d691db02a1ae456a9cbbc75f8d055099 : FAILURE in 9m 42s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/823a9af520df4dfd8968951ce30ec862 : FAILURE in 6m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b802223fc5b4c488aa386804694ca6b : FAILURE in 9m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b9c4d6de80948698fe2fe7a52a8fae1 : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ff3a5eb754ffc56cad7043074da035a702b099f6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-05-06 15:51:24.000000000","message":"Patch Set 1: Code-Review-1\n\nYou need to remove the \".\" files as well with exception of .gitreview.","accounts_in_message":[],"_revision_number":1},{"id":"79c33e5275ea7b5d479e11b506ddd2965e526979","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-05-06 15:51:48.000000000","message":"Patch Set 1:\n\nI don\u0027t see an README.rst, please add one.","accounts_in_message":[],"_revision_number":1},{"id":"d795d379140a5230fda6205dc17790aae74d9fa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-06 16:08:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5146f9bbd48d9bcf4874f8b4166fb84344646011","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-06 16:09:46.000000000","message":"Patch Set 2:\n\nSorry I forgot git add REAME.rst ...\nThe remaining .gitignore has been removed.","accounts_in_message":[],"_revision_number":2},{"id":"6182a771a3b06cffd756875dd73838a4431b5612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 16:13:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/ffab744b86b74e76a1749c1e65188a5e : FAILURE in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e241e04c86b05b6a3911cd1893d7a3a23cbc64e7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 08:04:24.000000000","message":"Topic retire-glare removed","accounts_in_message":[],"_revision_number":2},{"id":"014c272ff9ff3c789bbba4ea96c7c24b4adb969f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-08 08:04:27.000000000","message":"Topic set to retire-puppet-glare","accounts_in_message":[],"_revision_number":2},{"id":"c16e80785e055403b9a822b309aadd86d6269994","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-10 19:51:53.000000000","message":"Patch Set 2: Code-Review+1\n\nonce depends-on merge, gate will be green here.","accounts_in_message":[],"_revision_number":2},{"id":"f9a11c014c4ad2efa041e6e34f72b2779dd97233","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-05-11 14:58:20.000000000","message":"Patch Set 2:\n\nrecheck roject-config patch merged","accounts_in_message":[],"_revision_number":2},{"id":"1d53c897bb4d078380fa1096c0245a1f1fe59745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 14:58:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/b2406a284a2a44db90b06514d32d8d2b : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"ba9adade283842ed097a25a096570495d0db5f9f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 15:09:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nWe already received approval from three core reviewers in reviews and ML.\nSince the step 1 patch has been merged I\u0027ll approve this now.","accounts_in_message":[],"_revision_number":2},{"id":"917c73ff3a9d6ef18153e3abb720e04e5cb64fe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:10:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b01246e062aab0863da9cbac3a68cef8fe0c9596","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:11:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/7215b88cf5544a75b2bd180ad16d783d : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"a34be2bdb588d59155bf50eea029034a55ee51fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:11:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ba81a153712f1ebd6f79c94f28c62eb2f04090f9","revisions":{"20790ff15855d5045197dd3b93166ce11f755881":{"kind":"REWORK","_number":1,"created":"2021-05-06 14:55:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/790081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/81/790081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/81/790081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/81/790081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/81/790081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/81/790081/1"}}},"commit":{"parents":[{"commit":"f6c156be722eac9bf3ab403ab04688700798ae59","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/f6c156be722eac9bf3ab403ab04688700798ae59"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:55:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:55:33.000000000","tz":540},"subject":"Retire puppet-glare - Step 2: Remove Project Content","message":"Retire puppet-glare - Step 2: Remove Project Content\n\nDepends-on: https://review.opendev.org/790056\nChange-Id: Id76e00fafd171f3d77b70817c5cfd103cb207b0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/20790ff15855d5045197dd3b93166ce11f755881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/20790ff15855d5045197dd3b93166ce11f755881"}]},"branch":"refs/heads/master"},"ba81a153712f1ebd6f79c94f28c62eb2f04090f9":{"kind":"REWORK","_number":2,"created":"2021-05-06 16:08:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/790081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-glare","ref":"refs/changes/81/790081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/81/790081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/81/790081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-glare refs/changes/81/790081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-glare refs/changes/81/790081/2"}}},"commit":{"parents":[{"commit":"f6c156be722eac9bf3ab403ab04688700798ae59","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/f6c156be722eac9bf3ab403ab04688700798ae59"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 14:55:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-06 16:08:33.000000000","tz":540},"subject":"Retire puppet-glare - Step 2: Remove Project Content","message":"Retire puppet-glare - Step 2: Remove Project Content\n\nDepends-on: https://review.opendev.org/790056\nChange-Id: Id76e00fafd171f3d77b70817c5cfd103cb207b0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/ba81a153712f1ebd6f79c94f28c62eb2f04090f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-glare/commit/ba81a153712f1ebd6f79c94f28c62eb2f04090f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
