)]}'
{"id":"openstack%2Fpuppet-gnocchi~654191","triplet_id":"openstack%2Fpuppet-gnocchi~master~Iae4bc60a54e9b4c1a1c87276629496e0b91abc9b","project":"openstack/puppet-gnocchi","branch":"master","topic":"move-to-opendev","hashtags":[],"change_id":"Iae4bc60a54e9b4c1a1c87276629496e0b91abc9b","subject":"Move to opendev","status":"MERGED","created":"2019-04-21 06:45:27.000000000","updated":"2019-04-22 18:06:22.000000000","submitted":"2019-04-22 18:06:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654191-1555956382177-a663f06a","meta_rev_id":"54c309616b392aa95b97d8818b5e6c6b7397b99a","_number":654191,"virtual_id_number":654191,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-22 18:06:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-04-22 13:24:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-22 16:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-04-22 16:55:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 13:24:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-22 16:55:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-22 18:06:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cae90ab5759d961c164dce09fb594954b5169fd5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 06:45:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59d56732235a230568f771d6e0f216c498729dd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 07:47:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/91/654191/1/check/puppet-openstack-lint-ubuntu-bionic/3322c8b/ : SUCCESS in 5m 10s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/91/654191/1/check/puppet-openstack-syntax-4-ubuntu-bionic/c2421a5/ : SUCCESS in 4m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/91/654191/1/check/puppet-openstack-syntax-5-ubuntu-bionic/4a317cb/ : SUCCESS in 5m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/91/654191/1/check/puppet-openstack-syntax-6-ubuntu-bionic/d6c7582/ : SUCCESS in 4m 49s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/91/654191/1/check/puppet-openstack-unit-4.8-centos-7/da5cc55/ : SUCCESS in 19m 33s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/91/654191/1/check/puppet-openstack-unit-5.5-centos-7/60a8a9f/ : SUCCESS in 17m 07s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/91/654191/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/551d39a/ : SUCCESS in 8m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/91/654191/1/check/puppet-openstack-unit-latest-ubuntu-bionic/e25c237/ : FAILURE in 8m 44s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/91/654191/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/2723381/ : FAILURE in 31m 05s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/91/654191/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/70f9507/ : SUCCESS in 53m 31s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/91/654191/1/check/puppet-openstack-beaker-centos-7/6e8d552/ : SUCCESS in 18m 54s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/91/654191/1/check/puppet-openstack-beaker-ubuntu-bionic/aa30fce/ : FAILURE in 11m 21s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/91/654191/1/check/build-openstack-releasenotes/c7fe5a1/html/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3f763b8ae8615fa8cace5daad2dba956fffbf38f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-22 13:24:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06841aba071a94345b6d2a2dec2eb720c513c152","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-22 16:55:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"796a7142df4aaf02b02e25037b1d40742d515c13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:56:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"244809eb35abc3fff7734767f465b78a33264d7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:06:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/91/654191/1/gate/puppet-openstack-lint-ubuntu-bionic/9487477/ : SUCCESS in 4m 55s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/91/654191/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/3340480/ : SUCCESS in 4m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/91/654191/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/c0ef806/ : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/91/654191/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/f40e58d/ : SUCCESS in 6m 58s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/91/654191/1/gate/puppet-openstack-unit-4.8-centos-7/ce3e4cd/ : SUCCESS in 11m 16s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/91/654191/1/gate/puppet-openstack-unit-5.5-centos-7/90d18ff/ : SUCCESS in 11m 17s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/91/654191/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/fb4ad7c/ : SUCCESS in 10m 25s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/91/654191/1/gate/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/3b03819/ : FAILURE in 32m 00s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/91/654191/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/608f685/ : SUCCESS in 1h 03m 05s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/91/654191/1/gate/puppet-openstack-beaker-centos-7/b814a92/ : SUCCESS in 16m 29s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/91/654191/1/gate/puppet-openstack-beaker-ubuntu-bionic/c759bca/ : FAILURE in 14m 59s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/91/654191/1/gate/build-openstack-releasenotes/dd6120d/html/ : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0d6ca5ecbddf0a43a376b16bd92d14e0fb6532ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:06:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5dfc798eb8bdbda6d818ef02831eed998dbf3fa7","revisions":{"5dfc798eb8bdbda6d818ef02831eed998dbf3fa7":{"kind":"REWORK","_number":1,"created":"2019-04-21 06:45:27.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/91/654191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/91/654191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/91/654191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/91/654191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/91/654191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/91/654191/1"}}},"commit":{"parents":[{"commit":"c24ff4d59aebd3bbda307472be5e6bb323982922","subject":"Fix Repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/c24ff4d59aebd3bbda307472be5e6bb323982922"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:48.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:48.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: Iae4bc60a54e9b4c1a1c87276629496e0b91abc9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/5dfc798eb8bdbda6d818ef02831eed998dbf3fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/5dfc798eb8bdbda6d818ef02831eed998dbf3fa7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
