)]}'
{"id":"openstack%2Fpuppet-gnocchi~661217","triplet_id":"openstack%2Fpuppet-gnocchi~master~Ief9913dc664b73fd7effcc0c9775f374ff8ac782","project":"openstack/puppet-gnocchi","branch":"master","topic":"memcached-inet6-prefix","hashtags":[],"change_id":"Ief9913dc664b73fd7effcc0c9775f374ff8ac782","subject":"Prefix memcached_servers IPv6 address with `inet6:`","status":"ABANDONED","created":"2019-05-24 07:00:01.000000000","updated":"2019-05-24 13:59:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"86a49c94e77c685bc68e2b8a031b54d733e9d27f","_number":661217,"virtual_id_number":661217,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-24 08:03:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 08:03:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3517501250202bd661037f3c26f9a2258ef8a3b1","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-24 07:00:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b406e9ed0f729ea4c9d9cacbe7fe45779ea7d707","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 08:03:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/17/661217/1/check/puppet-openstack-lint-ubuntu-bionic/d794c72/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/17/661217/1/check/puppet-openstack-syntax-4-ubuntu-bionic/e47faa5/ : SUCCESS in 3m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/17/661217/1/check/puppet-openstack-syntax-5-ubuntu-bionic/b5a9a6f/ : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/17/661217/1/check/puppet-openstack-syntax-6-ubuntu-bionic/4e9b48d/ : SUCCESS in 4m 27s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/17/661217/1/check/puppet-openstack-unit-4.8-centos-7/dfe3798/ : SUCCESS in 12m 17s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/17/661217/1/check/puppet-openstack-unit-5.5-centos-7/3aa3578/ : SUCCESS in 18m 24s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/17/661217/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/1f086bf/ : SUCCESS in 14m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/17/661217/1/check/puppet-openstack-unit-latest-ubuntu-bionic/b7feef5/ : SUCCESS in 8m 25s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/17/661217/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/2643b8a/ : FAILURE in 29m 58s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/17/661217/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/32a305f/ : SUCCESS in 1h 01m 30s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/17/661217/1/check/puppet-openstack-beaker-centos-7/05b88a2/ : SUCCESS in 16m 00s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/17/661217/1/check/puppet-openstack-beaker-ubuntu-bionic/5e3e261/ : FAILURE in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6dcaa8a691d50688da47cc69588d023479093e0","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-24 13:59:52.000000000","message":"Abandoned\n\nWill patch keystone module only.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9f7b5116ef2ef294352771de9bffcb21b5cfd1b","revisions":{"d9f7b5116ef2ef294352771de9bffcb21b5cfd1b":{"kind":"REWORK","_number":1,"created":"2019-05-24 07:00:01.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/17/661217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/17/661217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/17/661217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/17/661217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/17/661217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/17/661217/1"}}},"commit":{"parents":[{"commit":"3854d2f5316232aec8702395d0d7b39cab41e905","subject":"Add openstackclient installation to the client class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/3854d2f5316232aec8702395d0d7b39cab41e905"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 06:59:05.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 06:59:05.000000000","tz":120},"subject":"Prefix memcached_servers IPv6 address with `inet6:`","message":"Prefix memcached_servers IPv6 address with `inet6:`\n\nDepends-On: https://review.opendev.org/661136\nChange-Id: Ief9913dc664b73fd7effcc0c9775f374ff8ac782\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d9f7b5116ef2ef294352771de9bffcb21b5cfd1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d9f7b5116ef2ef294352771de9bffcb21b5cfd1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
