)]}'
{"id":"openstack%2Fpuppet-gnocchi~672456","triplet_id":"openstack%2Fpuppet-gnocchi~master~I4b32b1b983b7f3fd41ab2ef15273a492f09a9a48","project":"openstack/puppet-gnocchi","branch":"master","topic":"prep-release-train-3","hashtags":[],"change_id":"I4b32b1b983b7f3fd41ab2ef15273a492f09a9a48","subject":"Prepare Train M3","status":"MERGED","created":"2019-07-24 08:20:28.000000000","updated":"2019-07-24 13:43:48.000000000","submitted":"2019-07-24 13:43:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672456-1563975828684-31ef7618","meta_rev_id":"f9aee45ab7cd7a74f0ed5a31b4f673ca83c3c9c9","_number":672456,"virtual_id_number":672456,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-24 13:43:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-07-24 13:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-07-24 13:31:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 13:31:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-24 13:43:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6038bde5cfcf062cd2b65d31aeb2131c1625f4d0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-24 08:20:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19c7aec2e3d8173fd16d7e47c9790a9896fba249","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 09:13:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/56/672456/1/check/build-openstack-puppet-tarball/18dfda0/ : SUCCESS in 3m 02s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/56/672456/1/check/puppet-openstack-lint-ubuntu-bionic/8ba541b/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"637031915e1dafda80d59999ad30aad226e82e69","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-24 13:31:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6cce1f224a8f19bd1272308b4b70eed29ee24228","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:31:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24af2b81937457a57f92d956ffaffaec0d6858fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:43:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f898460f4715aa8643eec9c15754e30c5be6e569","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:43:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/56/672456/1/gate/build-openstack-puppet-tarball/8ccc4bf/ : SUCCESS in 2m 33s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/56/672456/1/gate/puppet-openstack-lint-ubuntu-bionic/92b3c1f/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d012e8135bc9dd61167d8ca1493b467e8090a7b","revisions":{"7d012e8135bc9dd61167d8ca1493b467e8090a7b":{"kind":"REWORK","_number":1,"created":"2019-07-24 08:20:28.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/56/672456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/56/672456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/672456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/672456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/672456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/672456/1"}}},"commit":{"parents":[{"commit":"9ae01ae5ee922206a1fe08a39e951b8f6632228b","subject":"Manage oslo_middleware/max_request_body_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/9ae01ae5ee922206a1fe08a39e951b8f6632228b"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-24 08:19:35.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-24 08:19:35.000000000","tz":480},"subject":"Prepare Train M3","message":"Prepare Train M3\n\nUpdate the version metadata for Train milestone 3\n\nChange-Id: I4b32b1b983b7f3fd41ab2ef15273a492f09a9a48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7d012e8135bc9dd61167d8ca1493b467e8090a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7d012e8135bc9dd61167d8ca1493b467e8090a7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
