)]}'
{"id":"openstack%2Fpuppet-gnocchi~677044","triplet_id":"openstack%2Fpuppet-gnocchi~stable%2Fqueens~Iad83e8bbe3904fe4bc25570d296b804ce692161c","project":"openstack/puppet-gnocchi","branch":"stable/queens","topic":"bug/1840206-stable/queens","hashtags":[],"change_id":"Iad83e8bbe3904fe4bc25570d296b804ce692161c","subject":"Configure ceph_secret as secret","status":"MERGED","created":"2019-08-17 01:24:49.000000000","updated":"2019-08-19 05:29:32.000000000","submitted":"2019-08-19 05:29:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677044-1566192572807-16970a7d","meta_rev_id":"97974ca03f05c08b6bda60859043dff203e562c4","_number":677044,"virtual_id_number":677044,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-08-19 05:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-08-18 19:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-08-19 03:56:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-18 19:37:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-08-19 05:29:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38313b3b35f0535dd519633dfac051bda56a2d5d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-17 01:24:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f46a37e6a60a5e98573688010073d075e65513c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 02:57:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/330c05ff8d34453eb1b103c0d0074953 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/16dad75b4461494e8370c937ff4cfc72 : SUCCESS in 4m 01s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/1165c2eba12c483eb0e8035ded9a068f : SUCCESS in 12m 46s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a74376892e6b495d8f798bab80afa118 : SUCCESS in 10m 09s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/123683ac4c944e44857b4dd546c774d3 : TIMED_OUT in 1h 31m 09s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/100e5eefe2144cef9dddf6e63c379568 : SUCCESS in 59m 35s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/44375dcebb384cffb6f11b9fed40e7c2 : SUCCESS in 17m 31s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/3e78d78dd689432cb4f33d0f56fe9be8 : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":1},{"id":"79c493e1636b822fd6fa55371afdb8c800a9dcf3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-18 19:37:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"73adfc20414f25a86bb963f09d2ff099e62a6c6d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-19 03:56:33.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b3b1d48ee40123ff14d8c7b1d679dc43be956c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 03:56:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61f7c79ae485f5e7b1f4dcab34161243d97da15a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 05:29:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/95f378b953c84cd3b6ccbe4e4a846a93 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/e22cc1ba7a8744d6abd10f9900219332 : SUCCESS in 5m 05s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/9be94b3576a345818fb9905c0a976ffe : SUCCESS in 18m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4ad538b1830a4f7c83b41a1e32666cf3 : SUCCESS in 17m 27s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/993f946ec2bf42029b3cf0fc6d431a39 : TIMED_OUT in 1h 31m 19s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/dad3b8476a38406d84e9a13e635689d6 : SUCCESS in 1h 01m 28s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c3c6781498a04838a0ebacf501965ddc : SUCCESS in 23m 03s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/31b270f1b50f45019f74d4681b7e7652 : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ab02e0bb8666b893c12a8ff891ec277bf0c44bcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 05:29:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f2715671268f0e3942147bb49147beed23a5159","revisions":{"1f2715671268f0e3942147bb49147beed23a5159":{"kind":"REWORK","_number":1,"created":"2019-08-17 01:24:49.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/44/677044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/44/677044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/44/677044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/44/677044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/44/677044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/44/677044/1"}}},"commit":{"parents":[{"commit":"f5356eee36f2ec2d20697f76604eb4800b3329ce","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/f5356eee36f2ec2d20697f76604eb4800b3329ce"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:14:08.000000000","tz":480},"committer":{"name":"zhongshengping","email":"chdzsp@163.com","date":"2019-08-17 01:24:49.000000000","tz":0},"subject":"Configure ceph_secret as secret","message":"Configure ceph_secret as secret\n\nParameter ceph_secret should be configured as a secret.\n\nChange-Id: Iad83e8bbe3904fe4bc25570d296b804ce692161c\nCloses-Bug: #1840206\n(cherry picked from commit 0d83a6e5c138b9085b30f397e37343f0a45690c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/1f2715671268f0e3942147bb49147beed23a5159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/1f2715671268f0e3942147bb49147beed23a5159"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
