)]}'
{"id":"openstack%2Fpuppet-gnocchi~687707","triplet_id":"openstack%2Fpuppet-gnocchi~stable%2Ftrain~Ic677b931bb8ec067e36bcf7b512ace697d7b53c6","project":"openstack/puppet-gnocchi","branch":"stable/train","hashtags":[],"change_id":"Ic677b931bb8ec067e36bcf7b512ace697d7b53c6","subject":"Switch to Train","status":"MERGED","created":"2019-10-09 23:05:00.000000000","updated":"2019-10-10 20:26:22.000000000","submitted":"2019-10-10 20:26:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687707-1570739182940-bad3269c","meta_rev_id":"1044284e3279e1e7d6f8919b03806c3be99cebba","_number":687707,"virtual_id_number":687707,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 20:26:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-10 19:15:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-10 19:15:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 19:15:11.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-10 20:26:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bdfbb5208f679de751019367b6e47a6b8e36a4d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-09 23:05:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32581cebcdca475daf746e4c7d9a7b83b67a091a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 00:03:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcfee96f55b04bdea198d4c1761cdc32 : SUCCESS in 6m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6948b76bbd1c44e48cf93fac19dc2c6e : SUCCESS in 4m 47s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/cb13116caf754881a43e0d319fdeb768 : SUCCESS in 14m 20s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2321f8abb3f040e8b6781ec54e7956cb : FAILURE in 32m 29s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/1d367797c4d94530888e85604ffb5dcc : SUCCESS in 54m 40s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/15d0e47ecac145e582abb75fe5c981bc : SUCCESS in 17m 26s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7a5a6ece7f141a085bb162cc065a510 : FAILURE in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6f39cc5dc01825ef3d08958ddeb05d2391a33ac","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-10 19:15:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4143e4fcb6d7b3670a48f6e58a5b621aede872d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 19:15:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f367cba0c272c8ecc991bc79e1f96f5ea5406683","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:26:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1044284e3279e1e7d6f8919b03806c3be99cebba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:26:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31bf14b02dfd46948d974a2c7d03e342 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1d19fa7ecef4b0089985113cab92023 : SUCCESS in 5m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/30783d9e7e784084bdfc7f8122558d6d : SUCCESS in 19m 49s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/aace0d03ef434a03a47a5b0a6d0817e7 : FAILURE in 36m 08s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a166f01c34ce445cba043a0a77b9c9c2 : SUCCESS in 1h 05m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/66e2736636ed4890b71425c2edb85380 : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68c577337e7f34e45cd987b15fb0f792b1c1c7db","revisions":{"68c577337e7f34e45cd987b15fb0f792b1c1c7db":{"kind":"REWORK","_number":1,"created":"2019-10-09 23:05:00.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/07/687707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/07/687707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/07/687707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/07/687707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/07/687707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/07/687707/1"}}},"commit":{"parents":[{"commit":"d28f554c307821c783a4d1fb6425290eff1784ba","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d28f554c307821c783a4d1fb6425290eff1784ba"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-09 23:04:51.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-09 23:04:51.000000000","tz":120},"subject":"Switch to Train","message":"Switch to Train\n\nChange-Id: Ic677b931bb8ec067e36bcf7b512ace697d7b53c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/68c577337e7f34e45cd987b15fb0f792b1c1c7db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/68c577337e7f34e45cd987b15fb0f792b1c1c7db"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
