)]}'
{"id":"openstack%2Fpuppet-gnocchi~688838","triplet_id":"openstack%2Fpuppet-gnocchi~stable%2Fqueens~I65ed470a1f585ed4704057a71511451c8db44f8c","project":"openstack/puppet-gnocchi","branch":"stable/queens","topic":"bug/1847941-stable/queens","hashtags":[],"change_id":"I65ed470a1f585ed4704057a71511451c8db44f8c","subject":"Configure s3_secret_access_key as secret","status":"MERGED","created":"2019-10-16 00:41:11.000000000","updated":"2019-10-16 07:07:18.000000000","submitted":"2019-10-16 07:07:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688838-1571209638083-fe29816b","meta_rev_id":"15db4bd1a231199a5c13c009d8b4305fe890cabd","_number":688838,"virtual_id_number":688838,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 07:07:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-10-16 05:35:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-10-16 05:35:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 07:07:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc020f7d675bf264341c55e8a61257a834a64332","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-16 00:41:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"88708297cf96445391790f1585be13cbaf7e5465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 02:17:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/2a40fa9d65d04c22ade74b3bd6e66c21 : SUCCESS in 4m 56s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/d688d92967244db6899f9b31aa5d0197 : SUCCESS in 5m 12s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/5eaea1884bc64f358593422479fedf28 : SUCCESS in 11m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f1d7c67c281d4da09466a669e759cce4 : SUCCESS in 8m 30s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/0a6f6ad45200474694574dcf979cf332 : TIMED_OUT in 1h 30m 55s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a45105d636254026bbae3e85a07250ff : SUCCESS in 1h 05m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/75c2f4238bee440bade4957431e11539 : SUCCESS in 17m 15s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/d5c50ac10182460a90b07fdb332eb136 : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5fb24fa05a6fc38f6bc7399e6b59e5eca8ab0be5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-16 05:35:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fe7e83eb8140888ba2e6db6aa5a4446d075e508","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 05:35:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64c950b099539c8244e42f753809a27a8aeb213e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 07:07:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/2ed596dc73b94437bc9918b4e4b65e77 : SUCCESS in 7m 55s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/baf8494ae39e4f1a98d67f6ac92273ce : SUCCESS in 4m 45s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/fe2d783c4bd34033b7871433e8bb5781 : SUCCESS in 14m 07s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/eaade42a757d4d2d86fbcca5e88da423 : SUCCESS in 14m 08s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial-luminous https://zuul.opendev.org/t/openstack/build/de38e37ed98a4f6c9eca762d9c997931 : TIMED_OUT in 1h 31m 05s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/90b278140f3e46c79341cb12f3cb9c6d : SUCCESS in 19m 57s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/6244ff382fc145cdbf5a3b0d5e45d9fb : SUCCESS in 15m 34s","accounts_in_message":[],"_revision_number":1},{"id":"960f1cc027ffe603c21699f7dcf9d8d5faea2b7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 07:07:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a165cb4fef34d8432a04456de1575d021cd677b","revisions":{"0a165cb4fef34d8432a04456de1575d021cd677b":{"kind":"REWORK","_number":1,"created":"2019-10-16 00:41:11.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/38/688838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/38/688838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/688838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/688838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/688838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/688838/1"}}},"commit":{"parents":[{"commit":"7c991657c46748716f701d018c5dbe54daa8a4d1","subject":"Add CORS config support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7c991657c46748716f701d018c5dbe54daa8a4d1"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-10-14 00:52:25.000000000","tz":480},"committer":{"name":"zhongshengping","email":"chdzsp@163.com","date":"2019-10-16 00:41:11.000000000","tz":0},"subject":"Configure s3_secret_access_key as secret","message":"Configure s3_secret_access_key as secret\n\nParameter s3_secret_access_key should be configured as a secret.\n\nChange-Id: I65ed470a1f585ed4704057a71511451c8db44f8c\nCloses-Bug: #1847941\nDepends-On: https://review.opendev.org/688319\n(cherry picked from commit 05ee0bb4339b98ab67e4bedc192748905afe0aba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/0a165cb4fef34d8432a04456de1575d021cd677b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/0a165cb4fef34d8432a04456de1575d021cd677b"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
