)]}'
{"id":"openstack%2Fpuppet-gnocchi~713448","triplet_id":"openstack%2Fpuppet-gnocchi~master~I88dff282df4ce477e543dd2fcf5052a9ac472b84","project":"openstack/puppet-gnocchi","branch":"master","topic":"storage-deprecated","hashtags":[],"change_id":"I88dff282df4ce477e543dd2fcf5052a9ac472b84","subject":"Deprecate parameters under gnocchi::storage","status":"MERGED","created":"2020-03-17 13:55:18.000000000","updated":"2020-03-20 23:28:05.000000000","submitted":"2020-03-20 23:26:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713448-1584746802486-b16e9fe7","meta_rev_id":"71880b3f772ac85763a55f3adfc68709420501c7","_number":713448,"virtual_id_number":713448,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 23:26:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-20 22:14:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-03-20 22:19:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-03-20 22:19:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-17 22:00:15.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-20 22:14:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-20 22:19:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-20 23:26:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5abda03c3e4572b1c426b38c459e2e9cebb0e708","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-17 13:55:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56a59633ee1c33027ec7f1d4f2e22d7d3520284d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 15:05:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c316c4314b46408b844859c0c07d41cd : SUCCESS in 6m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32b31cea56f845fc8f07a86861702eb6 : SUCCESS in 5m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a74a80600bfd4392bb7d9806085e2372 : SUCCESS in 6m 06s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ad5784872baf471cb3ae9c70437d50c4 : SUCCESS in 20m 04s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/509381fc7125495d800837cf618b458c : SUCCESS in 14m 26s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/fe1c672178394c29935ec3d8a2d8fd10 : SUCCESS in 15m 24s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ee1f9e58c1d4b20add3ab2de8535be9 : SUCCESS in 10m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35892d3da37b408ba46c2274bc2b711f : SUCCESS in 13m 45s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0d0d6d79ed624409959b537ec4bb08fd : SUCCESS in 48m 42s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/61cd2a00a02b44ca97668d9d2deabdfb : SUCCESS in 56m 47s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5848cb5e433c426a8ba62c48eedbb095 : SUCCESS in 1h 06m 57s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a2113d1e15fe4267bc982fefaae53caf : SUCCESS in 21m 46s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09c4e5521a9d47c3afb7aad8c37f314d : FAILURE in 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60845b72c54a70d3b6bdd97cf460aeba4af8ebc9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-17 22:00:15.000000000","message":"Patch Set 1:\n\nshould we deprecate it in storage class and move it out to init or smth else?","accounts_in_message":[],"_revision_number":1},{"id":"e87183a6573ae94fd84e4f2d937919af9b7f60d4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-18 00:45:41.000000000","message":"Patch Set 1: Workflow-1\n\nThanks, Tobias! It makes sense to have the parameter in init as you suggested.\nLet me put WIP until I\u0027ll address that.","accounts_in_message":[],"_revision_number":1},{"id":"785d1b82c68027c6b50de81729bd5d554f83a5ec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-18 12:59:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0fc7893934f743e57949651be0891c5844cb4408","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-18 13:42:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"012a6400e9b784eaa84fa6390e83723c689f65a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 14:52:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab2b179e1ab34f879e319dd967305320 : FAILURE in 6m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/faf08fba6de542acb5119feae65ffc94 : SUCCESS in 6m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/301b6a8981304ac0b103f3e62a1e2cdd : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/114ad6acbc7a4ac482921a4b2d82a0b3 : FAILURE in 10m 43s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/46226b3b56144aa28eb3b914904b6312 : FAILURE in 10m 43s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/869612cd177a43dda8c31ac791cf079d : FAILURE in 11m 02s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4639c98865e94eabbcb709ba3ab04351 : FAILURE in 8m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4165727437d4ddcab61ad138dbe8833 : FAILURE in 9m 50s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ce127959f0d442f7b43550185118a292 : SUCCESS in 46m 22s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/215bf13ef84941b3b623912db7b2be0b : SUCCESS in 58m 18s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9248cf6d72074fba84c0b255a69b24bf : SUCCESS in 1h 01m 37s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/85e7f207f3604276b404d3e3f7d6ee6a : SUCCESS in 18m 30s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8affe2946c41434b8867359c7f8d897b : FAILURE in 13m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58e732f1eaaa473a8869f5c44c0125f2 : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"10e891f5a25bafbf3b87ac3b8efe4a7a877b5f77","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-18 23:00:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"232f9a2b3771bd9533646d90e2ea52e7761cff01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 23:24:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/323def6da6c14da1be08272e0a7c9af5 : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1be0dd1e517f47dc942bf69e4fddb333 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26ccb61e4f2949aba4539a89f5d63bf0 : SUCCESS in 4m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6efde81a27db44d4885ac3149e8746d1 : FAILURE in 9m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7ab2d9a14f2c444b8c054849264451ef : FAILURE in 9m 38s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/36dfa9521b7e41b98871bec76827d38c : FAILURE in 10m 30s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc091b9bfcfe4e53bb6a451f2ff98565 : FAILURE in 8m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72e2eb9d57ef45ae93c4b21b1a0b3226 : FAILURE in 8m 03s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/605b36114f6d45b6bc4c56c6fda750e8 : FAILURE in 6m 09s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/1a013866dd3747dda09aaf4367f55eb6 : FAILURE in 8m 33s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6b9602f3a29f4183892567cf37355564 : FAILURE in 13m 16s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fbca5b0919304e25bc61645cc24c9c59 : FAILURE in 8m 52s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5cfd83b8cab4619887bdd69d9ba065d : FAILURE in 6m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49bf486182b74c1fbbe59a3900e8b7ee : FAILURE in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"be7a47cb21507e1077245dfac2045bd390491dbc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-19 12:57:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5c7f707e43f7fd3465a946dc7d58ce10311c71ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 13:21:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56ef094cd32e4e0e8f570ceb8a3ada75 : SUCCESS in 5m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a78043ce29d24ec6b03b5a542ab98cda : SUCCESS in 7m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9b33c27af0f416f80be94bf7c4eec51 : SUCCESS in 7m 31s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/64543653813f4e4886464a45ee1a2843 : FAILURE in 15m 00s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/13896ae80c1d4edbb7d235971db9b707 : FAILURE in 18m 10s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2e9094fd8e5e442ea2c1652b4be82a57 : FAILURE in 15m 45s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4be24c053e3a41e0a37bd171005912c3 : FAILURE in 13m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52d9c0da5cac463c9623ce0525c0d3d6 : FAILURE in 10m 58s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f9a332efcf154f979bc1ba71e8dd3c0b : FAILURE in 10m 47s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/2f42584c3bc64fcab98fc79e94373a37 : FAILURE in 10m 32s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5efdf46f15344bd993c2fb481d019a86 : FAILURE in 11m 42s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ec461519e0304742a97af038f40ade0a : FAILURE in 11m 27s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bcd144219a341bf902ce0308a605236 : FAILURE in 8m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f618932a04cd40a9829e6fb1d5270c16 : FAILURE in 3m 50s","accounts_in_message":[],"_revision_number":4},{"id":"ad08a7bf733fca65813904bc5bdc118c115bec4a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-19 13:25:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"75989df0b1aa02c7012b1314452f584ee98b8fee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 13:51:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed4fad567aee44cbab03440303048a51 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf371397a1e24498a2beebd0fc89eed4 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6fead56855e429dbf3ea9e2f82466bf : SUCCESS in 5m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3ee90cd9907145f9b5c539c591032879 : SUCCESS in 20m 01s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/813debaa329c4cf0ae0181f9fd2fed6c : SUCCESS in 11m 04s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2f3f74454e7a422a83f56819e6a3cde7 : SUCCESS in 18m 07s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d444d10ab3d45b2b0676565a647f4f2 : SUCCESS in 21m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d6ccb2abfe741da81feee006f8724ce : SUCCESS in 11m 17s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7704d42bc3464203953ba43ead38c709 : FAILURE in 9m 13s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/db2ae6f54a2b4f9cb7ca65c8a94caad3 : FAILURE in 11m 09s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d8906ba163884d8fa7e1fbbe3e7b0905 : FAILURE in 12m 54s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c9d18fcd5e5c4ff8bf1c5df3560089cf : FAILURE in 12m 16s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2df0afc79a9a439eb75da340c0926631 : FAILURE in 7m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33353fde2bb041a8b79cf9b5abe02374 : FAILURE in 6m 39s","accounts_in_message":[],"_revision_number":5},{"id":"398dfacb748f3b6f84752c8283c0f23607cc6823","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-19 14:30:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"515277b696463eb837004a9ade785109d4cf842b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 15:38:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/219f253d2b394342ac3c86073ca8e027 : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cca57f080b4b473d9453acd516b04cf6 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64615154b1894c9ba07d0802aafd78b4 : SUCCESS in 5m 16s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/eab7178921db4c1db988f8596eed6cc3 : SUCCESS in 12m 05s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8870781074054736ae2b324a9a807a24 : SUCCESS in 10m 43s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/7fe600ec220a444fac866468dca0f359 : SUCCESS in 12m 24s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58fe0fd057c14419acc8737892c8248d : SUCCESS in 10m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4292c98cd1b4e73b2cc97215d26f39c : SUCCESS in 9m 57s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/443b29fe7f70465d9cf7638b0aad4c3e : SUCCESS in 50m 17s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/e6b2c6de491e47bfa058db202b9aa2e5 : SUCCESS in 1h 03m 04s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/69ccc9431fe849db8e2946495e361b97 : SUCCESS in 56m 54s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ac277373fd6342b5afffdfca141d33d3 : SUCCESS in 20m 29s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a22ea668d8c4a0f96573aadbfeb3f31 : FAILURE in 7m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cfab59454b1484a83166ba8c8e45884 : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":6},{"id":"ee6fbb9a205304413e319cd7acd1913f8453fac3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-20 13:07:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"40765de50b21d2b8f050b12fdb354edf8e23f187","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 14:35:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4543647c9b947129d02c01f9661990a : SUCCESS in 5m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7faa581d9b6d4640af08a8d11ef1de62 : SUCCESS in 6m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f36792757074245881b9fc941c5c042 : SUCCESS in 5m 35s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/05a3ac4e491e4f33a892d798466c456d : SUCCESS in 19m 00s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/861ef945969044eb9326caaf83589140 : SUCCESS in 13m 30s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a4e5deccfb0b4fa0b2687c6381c3743e : SUCCESS in 14m 20s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1192d93f907c4294a0b554ac9a1df1e3 : SUCCESS in 9m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47bc43a25a594208a047add514af0fa1 : SUCCESS in 10m 06s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/912bfd0cba0a461c83d282454e1ea3b4 : SUCCESS in 46m 08s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/4f017b05b8094217a1971a47e78326f2 : SUCCESS in 1h 14m 36s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aa7a709a14f94eb99090a10a8b194653 : SUCCESS in 1h 15m 08s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a7a1529663fa45d098f320d08cd768f2 : SUCCESS in 17m 54s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/319887b6acac46d0ab124c50cc5db28b : FAILURE in 8m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a82968bcf87446beb56d54b7e9772913 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":7},{"id":"2d582f40d3731c0db66271bbde751e6f7b0d4577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:04:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62065368d65b4331a13fbfe2456f0a39 : SUCCESS in 6m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ca63b9a2ce84860b06bfddf43686ab3 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a26f5e4df1024868b25abce1512f0227 : SUCCESS in 4m 52s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b5c5a79c6359495a9a81900148419ef3 : SUCCESS in 19m 27s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ac974c856b344dc0b592f197a2bd1870 : SUCCESS in 13m 20s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/b7680fcced7a449ba11776cfef080a30 : SUCCESS in 16m 58s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da60e3f32d8c4a4499bd05e81618dba6 : SUCCESS in 12m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fc30c924ae748e1aa2f50f9f50afec9 : SUCCESS in 12m 39s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/393412ee89c947ee9b5afd9fa9d89d8e : SUCCESS in 48m 48s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/8cf51a05c0594f298aab6e76b35224bf : SUCCESS in 1h 17m 56s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eace5e04f469473da4547793bed3059e : SUCCESS in 1h 08m 21s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/802b810a6c5b440f8fb4b4e4d561d936 : SUCCESS in 21m 00s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd06406a1ff94973b80c6c706a2ddfe5 : FAILURE in 10m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb22ef924d1645e7946266d881212f2f : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":7},{"id":"77377726fa046322c9538253733cdc11d9267c8f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-20 22:14:48.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"8b7050321087a9fcd96a1688cd1160e329019fa7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-20 22:19:21.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"2fab3be51fd7c18a82e0cf3f86d98b99d4cbd4a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 22:19:55.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0af86d1c9ba407005743d4b5449bea17ba3f11a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 23:26:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"402895b1b66e6502292e7cae849abd5fafb19895","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 23:26:42.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d61c9c7397d4a6e90e9b6ae452bdd66 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/defc719f4d91447d9c5996fc6edd3a86 : SUCCESS in 5m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfec4c69640046ed834f4fed50175471 : SUCCESS in 5m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/895b7a5a19954f7c95a80a248d34d04d : SUCCESS in 20m 29s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9db6be61e0844bcca351205e5c6bf890 : SUCCESS in 10m 14s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/380965ac7af9447ea086705f1f1852f5 : SUCCESS in 46m 56s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/82a2539115d64ef19aa88ef5e1917068 : SUCCESS in 1h 00m 29s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/299e6d909c174a60b484521fd4498ba7 : SUCCESS in 19m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb8f872b01b64c6eae14ae7ebf6789f1 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":7},{"id":"06c2a5b7e0b42f18116061416f5534b91598f0ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 23:28:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aecb8676d07446b485c9d8e145d7e954 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"271ccb6103bab6db6a08d9f8cace803b969d699a","revisions":{"c58f417dc41939bdc1f4452106ab78de6a86a372":{"kind":"REWORK","_number":1,"created":"2020-03-17 13:55:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/713448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/48/713448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/1"}}},"commit":{"parents":[{"commit":"036d7b5fe28ce51d78c58f1124e6bdfa9147b751","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/036d7b5fe28ce51d78c58f1124e6bdfa9147b751"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-17 13:54:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-17 13:55:16.000000000","tz":540},"subject":"Set coordination_url under DEFAULT section","message":"Set coordination_url under DEFAULT section\n\n... because the one in storage section was deprecated in gnocchi[1] .\n\n[1] https://github.com/gnocchixyz/gnocchi/commit/70b9ca427ba7710f5b45dda0b2e1490c08219f37\n\nChange-Id: I88dff282df4ce477e543dd2fcf5052a9ac472b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/c58f417dc41939bdc1f4452106ab78de6a86a372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/c58f417dc41939bdc1f4452106ab78de6a86a372"}]},"branch":"refs/heads/master"},"ce7652607522de3425dc16ad05273477e1be0d15":{"kind":"REWORK","_number":2,"created":"2020-03-18 13:42:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/713448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/48/713448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/2"}}},"commit":{"parents":[{"commit":"036d7b5fe28ce51d78c58f1124e6bdfa9147b751","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/036d7b5fe28ce51d78c58f1124e6bdfa9147b751"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:37:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:42:19.000000000","tz":540},"subject":"Deprecate parameters under gnocchi::storage","message":"Deprecate parameters under gnocchi::storage\n\ncoordination_url[1] and metric_processing_delay[2] under storage\nsection are both deprecated, so deprecate parameters under\ngnocchi::storage and migrate these parameters to appropriate\nclasses, gnocchi and gnocchi::metricd .\n\n[1] https://github.com/gnocchixyz/gnocchi/commit/70b9ca427ba7710f5b45dda0b2e1490c08219f37\n[2] https://github.com/gnocchixyz/gnocchi/commit/72fdba704d8b3862b2b0adfb0e38c8429149c1b0\nChange-Id: I88dff282df4ce477e543dd2fcf5052a9ac472b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/ce7652607522de3425dc16ad05273477e1be0d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/ce7652607522de3425dc16ad05273477e1be0d15"}]},"branch":"refs/heads/master"},"12c889f1c0124d8cc9beb827bc09da84e4a82673":{"kind":"REWORK","_number":3,"created":"2020-03-18 23:00:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/713448/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/48/713448/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/3"}}},"commit":{"parents":[{"commit":"036d7b5fe28ce51d78c58f1124e6bdfa9147b751","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/036d7b5fe28ce51d78c58f1124e6bdfa9147b751"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:37:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 23:00:24.000000000","tz":540},"subject":"Deprecate parameters under gnocchi::storage","message":"Deprecate parameters under gnocchi::storage\n\ncoordination_url[1] and metric_processing_delay[2] under storage\nsection are both deprecated, so deprecate parameters under\ngnocchi::storage and migrate these parameters to appropriate\nclasses, gnocchi and gnocchi::metricd .\n\n[1] https://github.com/gnocchixyz/gnocchi/commit/70b9ca427ba7710f5b45dda0b2e1490c08219f37\n[2] https://github.com/gnocchixyz/gnocchi/commit/72fdba704d8b3862b2b0adfb0e38c8429149c1b0\nChange-Id: I88dff282df4ce477e543dd2fcf5052a9ac472b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/12c889f1c0124d8cc9beb827bc09da84e4a82673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/12c889f1c0124d8cc9beb827bc09da84e4a82673"}]},"branch":"refs/heads/master"},"451e0e2da6a3f053255e9bcd963b96818bdaaf44":{"kind":"REWORK","_number":4,"created":"2020-03-19 12:57:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/713448/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/48/713448/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/4"}}},"commit":{"parents":[{"commit":"036d7b5fe28ce51d78c58f1124e6bdfa9147b751","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/036d7b5fe28ce51d78c58f1124e6bdfa9147b751"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:37:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-19 12:57:14.000000000","tz":540},"subject":"Deprecate parameters under gnocchi::storage","message":"Deprecate parameters under gnocchi::storage\n\ncoordination_url[1] and metric_processing_delay[2] under storage\nsection are both deprecated, so deprecate parameters under\ngnocchi::storage and migrate these parameters to appropriate\nclasses, gnocchi and gnocchi::metricd .\n\n[1] https://github.com/gnocchixyz/gnocchi/commit/70b9ca427ba7710f5b45dda0b2e1490c08219f37\n[2] https://github.com/gnocchixyz/gnocchi/commit/72fdba704d8b3862b2b0adfb0e38c8429149c1b0\nChange-Id: I88dff282df4ce477e543dd2fcf5052a9ac472b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/451e0e2da6a3f053255e9bcd963b96818bdaaf44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/451e0e2da6a3f053255e9bcd963b96818bdaaf44"}]},"branch":"refs/heads/master"},"5d8de592c728d846a7e9c23fba20bf19e5bf14db":{"kind":"REWORK","_number":5,"created":"2020-03-19 13:25:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/713448/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/48/713448/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/5"}}},"commit":{"parents":[{"commit":"036d7b5fe28ce51d78c58f1124e6bdfa9147b751","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/036d7b5fe28ce51d78c58f1124e6bdfa9147b751"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:37:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-19 13:25:15.000000000","tz":540},"subject":"Deprecate parameters under gnocchi::storage","message":"Deprecate parameters under gnocchi::storage\n\ncoordination_url[1] and metric_processing_delay[2] under storage\nsection are both deprecated, so deprecate parameters under\ngnocchi::storage and migrate these parameters to appropriate\nclasses, gnocchi and gnocchi::metricd .\n\n[1] https://github.com/gnocchixyz/gnocchi/commit/70b9ca427ba7710f5b45dda0b2e1490c08219f37\n[2] https://github.com/gnocchixyz/gnocchi/commit/72fdba704d8b3862b2b0adfb0e38c8429149c1b0\nChange-Id: I88dff282df4ce477e543dd2fcf5052a9ac472b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/5d8de592c728d846a7e9c23fba20bf19e5bf14db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/5d8de592c728d846a7e9c23fba20bf19e5bf14db"}]},"branch":"refs/heads/master"},"f38f7182dec091cc6c3c3055d4d2f0b7ca6a20ba":{"kind":"REWORK","_number":6,"created":"2020-03-19 14:30:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/713448/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/48/713448/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/6"}}},"commit":{"parents":[{"commit":"036d7b5fe28ce51d78c58f1124e6bdfa9147b751","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/036d7b5fe28ce51d78c58f1124e6bdfa9147b751"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:37:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-19 14:29:54.000000000","tz":540},"subject":"Deprecate parameters under gnocchi::storage","message":"Deprecate parameters under gnocchi::storage\n\ncoordination_url[1] and metric_processing_delay[2] under storage\nsection are both deprecated, so deprecate parameters under\ngnocchi::storage and migrate these parameters to appropriate\nclasses, gnocchi and gnocchi::metricd .\n\n[1] https://github.com/gnocchixyz/gnocchi/commit/70b9ca427ba7710f5b45dda0b2e1490c08219f37\n[2] https://github.com/gnocchixyz/gnocchi/commit/72fdba704d8b3862b2b0adfb0e38c8429149c1b0\nChange-Id: I88dff282df4ce477e543dd2fcf5052a9ac472b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/f38f7182dec091cc6c3c3055d4d2f0b7ca6a20ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/f38f7182dec091cc6c3c3055d4d2f0b7ca6a20ba"}]},"branch":"refs/heads/master"},"271ccb6103bab6db6a08d9f8cace803b969d699a":{"kind":"REWORK","_number":7,"created":"2020-03-20 13:07:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/713448/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/48/713448/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/48/713448/7"}}},"commit":{"parents":[{"commit":"036d7b5fe28ce51d78c58f1124e6bdfa9147b751","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/036d7b5fe28ce51d78c58f1124e6bdfa9147b751"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:37:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-20 13:06:53.000000000","tz":540},"subject":"Deprecate parameters under gnocchi::storage","message":"Deprecate parameters under gnocchi::storage\n\ncoordination_url[1] and metric_processing_delay[2] under storage\nsection are both deprecated, so deprecate parameters under\ngnocchi::storage and migrate these parameters to appropriate\nclasses, gnocchi and gnocchi::metricd .\n\n[1] https://github.com/gnocchixyz/gnocchi/commit/70b9ca427ba7710f5b45dda0b2e1490c08219f37\n[2] https://github.com/gnocchixyz/gnocchi/commit/72fdba704d8b3862b2b0adfb0e38c8429149c1b0\nChange-Id: I88dff282df4ce477e543dd2fcf5052a9ac472b84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/271ccb6103bab6db6a08d9f8cace803b969d699a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/271ccb6103bab6db6a08d9f8cace803b969d699a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
