)]}'
{"id":"openstack%2Fpuppet-gnocchi~713656","triplet_id":"openstack%2Fpuppet-gnocchi~master~I6fb80bc63a9f7b370f9349681327a50d705d7514","project":"openstack/puppet-gnocchi","branch":"master","topic":"database_connection","hashtags":[],"change_id":"I6fb80bc63a9f7b370f9349681327a50d705d7514","subject":"Deprecate gnocchi::database_connection","status":"MERGED","created":"2020-03-18 13:50:40.000000000","updated":"2020-03-23 14:55:12.000000000","submitted":"2020-03-23 14:52:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713656-1584975145155-332a2ac9","meta_rev_id":"5cbec69f91b4a645e752af368b9d907d08785329","_number":713656,"virtual_id_number":713656,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 14:52:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-23 13:02:04.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-23 13:43:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-03-23 13:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-23 13:43:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 13:02:04.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-23 13:43:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-23 13:50:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-23 14:52:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1bbafaac632cc9bd497028b5cc1b5700310f3fe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-18 13:50:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe02172a213f3062cf67b81ca7063599fc28d54f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-18 13:51:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0c77ed43771027f1ba2fb69c2b67d19db4b1d776","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-18 13:52:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ea9c1e708edd2ed124ef06b1d980526411b94a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 15:03:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f3aefd3e0c345fcb03c6e67201cc0d7 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0066995342af43ee87d2977daddc06e6 : SUCCESS in 5m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f75e9c5c45c4cd6bca071d3b347a4d6 : SUCCESS in 5m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/cab61fd026da42eab04005d58b52530b : SUCCESS in 17m 57s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d05f31911f8447caa8fa4db368f705ba : SUCCESS in 11m 23s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/4794578814b547a6ab2b23fd92bcb841 : SUCCESS in 15m 29s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b05a895d45a84d8389928366b91f0cff : SUCCESS in 11m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40eddbccea4b4441af59fd85ba91a37a : SUCCESS in 11m 16s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/176f8fde07f7410c9c9c8884eb3b29cd : SUCCESS in 1h 01m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a079f5455d8d4516b793402af97484dc : SUCCESS in 1h 03m 14s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/66e9ea357aed45e5a89a7187edd93b05 : SUCCESS in 53m 36s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c69da504e82c42c9a18c643d68b0f75c : SUCCESS in 19m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16a7641e561949e88185fc2580a37cf0 : FAILURE in 18m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c00f1fdbebe4cd795eeeaf7b9fae555 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":3},{"id":"90217553064924ecadbbbe80187613fc1c846ccb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-20 22:13:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"94f01b73b5b98cc93391be3ecbb9ea57c8e9f87a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-20 22:19:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b05f6c48b97d5110822da178fcfb564a32aa536a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 23:26:42.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"2b7a77f876c1dd185ce92e21e4ebff58bb452264","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-22 13:03:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"94ae1968c9a8c788bfdb1ae8f91693709c334b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 14:13:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc095a828ca54a21841a807ffb1354b4 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c64a316890d472fac42618c6885c95e : SUCCESS in 4m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01ca339bb62341c48d0ef3e959461d42 : SUCCESS in 4m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/80ff1ba748674018a4344076e4054272 : SUCCESS in 11m 08s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7a31f67eeebc43f5b4952f13105b6632 : SUCCESS in 11m 30s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1b55e4a3cc844d369899eae2f822466e : SUCCESS in 14m 50s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7676929276e0464a9066348b4af27566 : SUCCESS in 14m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e12a051c9a5c4c9ea8ae6d1b98aeb17c : SUCCESS in 9m 21s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2edb4b4312c0463ab997569565df99d3 : SUCCESS in 53m 26s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/d6769ac2d3104efe91012cbb6781aa0a : SUCCESS in 1h 05m 42s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6dbbb4d709ec40eba22b10299d407d88 : SUCCESS in 1h 02m 58s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/782cf14be4434e3d86a7ead6a98de31b : SUCCESS in 20m 00s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24375d95f2844e9a8818969a905b3aad : FAILURE in 9m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3453794bc59644e9a49517b949f54460 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":4},{"id":"df21f3060a0c2423c889ea1c07173c8f25abb06b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-23 13:02:04.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"825ee5cf78b153aa079d39cc99aadb18c76440ab","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-23 13:43:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"53efd3282a0bb7af72a59f27874cf8f56c989871","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 13:44:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bf65062afad5d8020f310e2885dd807a4ded022a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-23 13:50:14.000000000","message":"Patch Set 4: Code-Review+2\n\nCan we follow up with a patch that removes the gnocchi::db include in init? That would make sure deployments doesn\u0027t break depending on the gnocchi vs gnocchi::db declaration order in manifests.","accounts_in_message":[],"_revision_number":4},{"id":"040a2ed26731bd708f01fb90a859d0a5fd9e36d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:52:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a05292891dd545c6bd9a7ddd9fe182e9 : SUCCESS in 6m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67b91f4f335a451e8269c9c9fe67f897 : SUCCESS in 5m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddd0dce7c80b4f78a780a72fd743995a : SUCCESS in 6m 08s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/62e03351b8c24638bcc49d57c71d1642 : SUCCESS in 11m 30s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b28a8c99c56d40c3903ba9ad3a821421 : SUCCESS in 13m 22s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b83950ec7a9d411c8e3283d34301edeb : SUCCESS in 47m 51s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a94c2d21bd7e493ba49530524dc5334b : SUCCESS in 1h 02m 29s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1ff90a564b5f4d12805d993f52cd4c82 : SUCCESS in 18m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af40e97ebfc7440bb2fe6a2544d2122d : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":4},{"id":"2fc5d40079bc5f6638badce8bebc2bb1169ee806","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:52:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a82ba6302c45ef234ccdc2309a5da692e8b80eea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:55:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b066c05ce1c64bd0a2b0f871a85ff27f : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b15098e93bd6a3ee0a01f01c8038812303d4b4ab","revisions":{"4a5d0ad16da92b266325faaf1a0a3b0c128ae7d9":{"kind":"REWORK","_number":1,"created":"2020-03-18 13:50:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/713656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/56/713656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/1"}}},"commit":{"parents":[{"commit":"036d7b5fe28ce51d78c58f1124e6bdfa9147b751","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/036d7b5fe28ce51d78c58f1124e6bdfa9147b751"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:47:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:50:32.000000000","tz":540},"subject":"Deprecate gnocchi::database_connection","message":"Deprecate gnocchi::database_connection\n\nSince gnocchi::db class was introcued[1], we have 2 parameters,\ngnocchi::database_connection and gnocchi::db::database_connection,\nto configure the same database_connection parameter.\n\nLet\u0027s deprecate the one in gnocchi class, so that we can have\nonly one parameter to set the one parameter.\n\n[1] https://github.com/openstack/puppet-gnocchi/commit/4d64bd45a79005a170b2407a5053356f2d5e09fb\nChange-Id: I6fb80bc63a9f7b370f9349681327a50d705d7514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/4a5d0ad16da92b266325faaf1a0a3b0c128ae7d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/4a5d0ad16da92b266325faaf1a0a3b0c128ae7d9"}]},"branch":"refs/heads/master"},"47fef2a7e5db59e6f2e271d2d06f36270d4486e6":{"kind":"REWORK","_number":2,"created":"2020-03-18 13:51:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/713656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/56/713656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/2"}}},"commit":{"parents":[{"commit":"036d7b5fe28ce51d78c58f1124e6bdfa9147b751","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/036d7b5fe28ce51d78c58f1124e6bdfa9147b751"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:47:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:51:18.000000000","tz":540},"subject":"Deprecate gnocchi::database_connection","message":"Deprecate gnocchi::database_connection\n\nSince gnocchi::db class was introcued[1], we have 2 parameters,\ngnocchi::database_connection and gnocchi::db::database_connection,\nto configure the same database_connection parameter.\n\nLet\u0027s deprecate the one in gnocchi class, so that we can have\nonly one parameter to set the one parameter.\n\n[1] https://github.com/openstack/puppet-gnocchi/commit/4d64bd45a79005a170b2407a5053356f2d5e09fb\nChange-Id: I6fb80bc63a9f7b370f9349681327a50d705d7514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/47fef2a7e5db59e6f2e271d2d06f36270d4486e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/47fef2a7e5db59e6f2e271d2d06f36270d4486e6"}]},"branch":"refs/heads/master"},"b5a205629c4e46f12e704dfeaa3029b5be9da839":{"kind":"REWORK","_number":3,"created":"2020-03-18 13:52:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/713656/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/56/713656/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/3"}}},"commit":{"parents":[{"commit":"036d7b5fe28ce51d78c58f1124e6bdfa9147b751","subject":"[ussuri][goal] Cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/036d7b5fe28ce51d78c58f1124e6bdfa9147b751"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:47:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:52:35.000000000","tz":540},"subject":"Deprecate gnocchi::database_connection","message":"Deprecate gnocchi::database_connection\n\nSince gnocchi::db class was introcued[1], we have 2 parameters,\ngnocchi::database_connection and gnocchi::db::database_connection,\nto configure the same database_connection parameter.\n\nLet\u0027s deprecate the one in gnocchi class, so that we can have\nonly one parameter to set the one parameter.\n\n[1] https://github.com/openstack/puppet-gnocchi/commit/4d64bd45a79005a170b2407a5053356f2d5e09fb\nChange-Id: I6fb80bc63a9f7b370f9349681327a50d705d7514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b5a205629c4e46f12e704dfeaa3029b5be9da839"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b5a205629c4e46f12e704dfeaa3029b5be9da839"}]},"branch":"refs/heads/master"},"b15098e93bd6a3ee0a01f01c8038812303d4b4ab":{"kind":"REWORK","_number":4,"created":"2020-03-22 13:03:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/713656/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/56/713656/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/56/713656/4"}}},"commit":{"parents":[{"commit":"271ccb6103bab6db6a08d9f8cace803b969d699a","subject":"Deprecate parameters under gnocchi::storage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/271ccb6103bab6db6a08d9f8cace803b969d699a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-18 13:47:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-22 13:03:06.000000000","tz":540},"subject":"Deprecate gnocchi::database_connection","message":"Deprecate gnocchi::database_connection\n\nSince gnocchi::db class was introcued[1], we have 2 parameters,\ngnocchi::database_connection and gnocchi::db::database_connection,\nto configure the same database_connection parameter.\n\nLet\u0027s deprecate the one in gnocchi class, so that we can have\nonly one parameter to set the one parameter.\n\n[1] https://github.com/openstack/puppet-gnocchi/commit/4d64bd45a79005a170b2407a5053356f2d5e09fb\nChange-Id: I6fb80bc63a9f7b370f9349681327a50d705d7514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b15098e93bd6a3ee0a01f01c8038812303d4b4ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b15098e93bd6a3ee0a01f01c8038812303d4b4ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
