)]}'
{"id":"openstack%2Fpuppet-gnocchi~721088","triplet_id":"openstack%2Fpuppet-gnocchi~master~Ide2bcd1326270b5ba0a8ce25f3388e8bf48a66a0","project":"openstack/puppet-gnocchi","branch":"master","topic":"keystone-v3","hashtags":[],"change_id":"Ide2bcd1326270b5ba0a8ce25f3388e8bf48a66a0","subject":"Add domain/region parameters for gnocchi::storage::swift","status":"MERGED","created":"2020-04-19 14:51:52.000000000","updated":"2020-04-20 20:57:58.000000000","submitted":"2020-04-20 20:56:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721088-1587416181102-5004bf0a","meta_rev_id":"3ab1bab11ec503c3fa829b21db8ea47da7794070","_number":721088,"virtual_id_number":721088,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 20:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-20 19:30:31.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-04-20 14:43:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-20 19:30:31.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 14:43:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-20 19:30:31.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-20 20:56:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1929846fdbd9fe72bc7d3833fcf82c3fa88a6b6a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-19 14:51:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbe5bf1427f83d2a5d35e1c8d2202697a2dc8f96","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-19 15:02:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5c49b6b33d2224b18f31ac1e38ef9e9071a6d367","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-19 15:21:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b2153845a3695e059f927c0a766cae14c6d4bcc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 16:26:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2e8a68d56b2478dab3ad818268d626a : SUCCESS in 6m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dde295fe3264a9db812034158a3e5d5 : SUCCESS in 6m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57d7d15370a64033a7aa0e3ae8309673 : SUCCESS in 4m 41s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/0846245a35064024b00df1a80ce92774 : SUCCESS in 13m 06s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/cf8cebe811a04f1997f58dbb5eabd049 : SUCCESS in 16m 03s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36209f22efe04803b1b74bb5d719a819 : SUCCESS in 12m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4937b76c318740789c1b544236369ce0 : SUCCESS in 9m 09s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5057d2d8102c4cd29ba44894eb66ce8f : SUCCESS in 1h 03m 08s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/22576baa44f1495283ae4db4815f4c87 : SUCCESS in 52m 19s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/70d9b86345c14b11a16834baaec3e50d : SUCCESS in 17m 17s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/629312b93c3745838c6fb759a85560ba : FAILURE in 14m 17s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2285f80a8e64885a4e4eef258e01899 : FAILURE in 6m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4cf6066abf343b3858097aa4d62182f : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"958b0be2dce2ddf8656deec67ab8e7b10da6b8c3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-20 14:43:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"207b83de754f6bbc0f4ae60c66bbdaaac5e07c6a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-20 19:30:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"19f4347d812951d03c82bc0cb9f697ef74d30cb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 19:30:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"092748d72e3419285c35e8733e82b2834d3c7325","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:56:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3308e6f71f4456a9eec4ba832ce0809 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e7f70a03e144062a6658a941b36d5ba : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c015deecb9534e53bfde5a9f21fc14df : SUCCESS in 4m 59s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5c82c5fd1993437e913e279fe4b59d8f : SUCCESS in 13m 55s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/baccad0b3c994664870621634c3bc9e6 : SUCCESS in 14m 34s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48d3b761920d41b6b529efb1e14caa4a : SUCCESS in 9m 46s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8badb53bdcfb4d7997736b87995e2691 : SUCCESS in 53m 28s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4a2e223af60143ec99b014c26c24b424 : SUCCESS in 1h 15m 22s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8eea48c22a6646caab33437e9f2df059 : SUCCESS in 19m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58dbcea24f554d5f817d3d16b93224b1 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":3},{"id":"5d45d155246f6f627561739e66063b271fde1ed2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:56:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ea9c2ff1ddc63d310b3294c40cf23417471e2064","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:57:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c9fec45dd81455ab3920d8038776389 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"237961239b818c3074688ce10b405c9ead124008","revisions":{"d6cf91de2e32f5b9a493f23b4280267482f91eb1":{"kind":"REWORK","_number":1,"created":"2020-04-19 14:51:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/721088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/88/721088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/1"}}},"commit":{"parents":[{"commit":"bfceb1e2dc675cd0e34bf893261faff80a9711b4","subject":"Use service default in gnocchi::storage::swift","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/bfceb1e2dc675cd0e34bf893261faff80a9711b4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-19 14:44:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-19 14:50:11.000000000","tz":540},"subject":"Add domain/region parameters for gnocchi::storage::swift","message":"Add domain/region parameters for gnocchi::storage::swift\n\n... so that operators can configure their keystone domain and region\nused for access to swift.\n\nChange-Id: Ide2bcd1326270b5ba0a8ce25f3388e8bf48a66a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d6cf91de2e32f5b9a493f23b4280267482f91eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/d6cf91de2e32f5b9a493f23b4280267482f91eb1"}]},"branch":"refs/heads/master"},"b80295ee5ce50d033aed3725e804f0356530dc27":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-19 15:02:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/721088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/88/721088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/2"}}},"commit":{"parents":[{"commit":"a130da928ab34ad91a0710ef1c627da786132d00","subject":"Use service default in gnocchi::storage::swift","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a130da928ab34ad91a0710ef1c627da786132d00"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-19 14:44:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-19 15:02:16.000000000","tz":540},"subject":"Add domain/region parameters for gnocchi::storage::swift","message":"Add domain/region parameters for gnocchi::storage::swift\n\n... so that operators can configure their keystone domain and region\nused for access to swift.\n\nChange-Id: Ide2bcd1326270b5ba0a8ce25f3388e8bf48a66a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b80295ee5ce50d033aed3725e804f0356530dc27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/b80295ee5ce50d033aed3725e804f0356530dc27"}]},"branch":"refs/heads/master"},"237961239b818c3074688ce10b405c9ead124008":{"kind":"REWORK","_number":3,"created":"2020-04-19 15:21:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/721088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/88/721088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/721088/3"}}},"commit":{"parents":[{"commit":"a130da928ab34ad91a0710ef1c627da786132d00","subject":"Use service default in gnocchi::storage::swift","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/a130da928ab34ad91a0710ef1c627da786132d00"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-19 14:44:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-19 15:21:38.000000000","tz":540},"subject":"Add domain/region parameters for gnocchi::storage::swift","message":"Add domain/region parameters for gnocchi::storage::swift\n\n... so that operators can configure their keystone domain and region\nused for access to swift.\n\nChange-Id: Ide2bcd1326270b5ba0a8ce25f3388e8bf48a66a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/237961239b818c3074688ce10b405c9ead124008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/237961239b818c3074688ce10b405c9ead124008"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
