)]}'
{"id":"openstack%2Fpuppet-gnocchi~728227","triplet_id":"openstack%2Fpuppet-gnocchi~master~I2ab815519b3c6116a7658d97218bb6eb18917697","project":"openstack/puppet-gnocchi","branch":"master","topic":"v-version-bump","hashtags":[],"change_id":"I2ab815519b3c6116a7658d97218bb6eb18917697","subject":"Bump version for the start of Victoria","status":"MERGED","created":"2020-05-14 18:26:07.000000000","updated":"2020-05-15 23:10:03.000000000","submitted":"2020-05-15 23:10:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728227-1589584203192-6f122f09","meta_rev_id":"02daeba7ee30851864dbf074ab245247c0ac6c7f","_number":728227,"virtual_id_number":728227,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 23:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-05-15 23:05:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-14 18:47:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-05-15 23:05:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 18:47:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-15 23:05:09.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-15 23:10:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc3e08ea5e7e1037552a25178617745a8cfecb74","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-14 18:26:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b190dfcd30418019567ecb44e3931bd13f9844d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:46:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/cd3973221bf74fc9a8f041fcc0dda9f2 : SUCCESS in 2m 32s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2d864df828c48d8b29ecb18761befc3 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8d5edebbe42f6a13185c2a8ceb991f39aef655c3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 18:47:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c51ae428d8fa333c1ea2b5a42db037a4c81b7a7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-15 23:05:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad4a667f2b16b938b15882e3f623c93c3ba1a76e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:05:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"889d27d7cf71515c4f9e1c302ca5b9c65c0fe008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:10:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/c89de5a9c1a041a29992751f53e24831 : SUCCESS in 2m 18s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79b60f27c5ec44a6ba6a406329744ac2 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"00d6ea6d83be311b99e15f9074a8ee2c42d50c7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:10:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f9181cc6b3dd8d6682037d7ad33ae49bef07c1c","revisions":{"5f9181cc6b3dd8d6682037d7ad33ae49bef07c1c":{"kind":"REWORK","_number":1,"created":"2020-05-14 18:26:07.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/27/728227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/27/728227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/728227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/728227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/728227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/27/728227/1"}}},"commit":{"parents":[{"commit":"db1b9c1d5c61e7c8802c9269d2c01edfcdf936cd","subject":"Merge \"[ussuri][goal] Add CONTRIBUTING.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/db1b9c1d5c61e7c8802c9269d2c01edfcdf936cd"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-05-14 18:24:27.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-05-14 18:26:07.000000000","tz":-360},"subject":"Bump version for the start of Victoria","message":"Bump version for the start of Victoria\n\nBump the major version for the Victoria release afte Ussuir has\nbeen branched.\n\nChange-Id: I2ab815519b3c6116a7658d97218bb6eb18917697\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/5f9181cc6b3dd8d6682037d7ad33ae49bef07c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/5f9181cc6b3dd8d6682037d7ad33ae49bef07c1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
