)]}'
{"id":"openstack%2Fpuppet-gnocchi~740788","triplet_id":"openstack%2Fpuppet-gnocchi~master~Iff924ab75d40ffba6a213c54a6e28933b18cf7e4","project":"openstack/puppet-gnocchi","branch":"master","topic":"740788","hashtags":[],"change_id":"Iff924ab75d40ffba6a213c54a6e28933b18cf7e4","subject":"Change to relative include","status":"MERGED","created":"2020-07-13 19:20:20.000000000","updated":"2020-07-14 09:09:11.000000000","submitted":"2020-07-14 09:09:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740788-1594717751032-6b2a4802","meta_rev_id":"67a5fdc3e382e78e4b0b2b94fb299b985acb2220","_number":740788,"virtual_id_number":740788,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-14 09:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-07-14 02:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-07-14 07:50:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-14 02:30:26.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-07-14 09:09:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3537de8ae3208753abf965ef55c9593b4cf9d098","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-13 19:20:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26bca193e39444a3cdbd50820dde4d5e4485a716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 20:55:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a50c3072043459dbd17b6b14a5dfcab : FAILURE in 6m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0cdcb015ce3465e9aba94701b5c57f3 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/238d601a8b5b48f08534eef9fef61315 : SUCCESS in 6m 23s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9d04812e0e744ab6bc0f164e046338e7 : SUCCESS in 12m 48s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/049fc9a579ed41f7aae7802d9cc05186 : SUCCESS in 24m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e9d1c57217340d4be9b75b0517200d1 : SUCCESS in 10m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75fab71efd9d499b9cc033de0306b44e : SUCCESS in 10m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/72e92cb1f1b14d22a95d610ff8164cd2 : SUCCESS in 50m 34s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2c5496b3c2ef4ed8997d416c016976a0 : SUCCESS in 1h 23m 49s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5d25c4a6d38d4a56b9d27b3eefa575bc : FAILURE in 8m 57s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9a03ed6cbf85443091cd74d4b55dbe90 : FAILURE in 11m 57s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5be4571ea0b042f7a9a89ee8f17e912e : FAILURE in 10m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"061882b1bead5e4cc02502cd2e63b0e9bda96474","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-13 21:44:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"217dd2aba24efb1b43afda9d9f4f6db10006150a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 22:38:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b124446ae2c4be192821bfe75fb92ee : SUCCESS in 4m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05f69924177e46f0aec8fbcf32c87f5f : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc92e45992c548a8a1ac9cd1f35d737c : SUCCESS in 4m 43s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/dcf18fa858394b4799b3f6ecf1260abd : SUCCESS in 9m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0435b664b98747a19e590a4243e8ebec : SUCCESS in 10m 17s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd6d0116b0754e868a781e95a937dd07 : SUCCESS in 9m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/076c2fef94ea43e5bdf37000ed354015 : SUCCESS in 9m 36s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/10fd4edb03a847938554136c55a87964 : SUCCESS in 35m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bd446befd58145e3a62da0f96e23c9f9 : SUCCESS in 51m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c8eeaafcb1414eb2ba5642eea2f679b5 : SUCCESS in 14m 41s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/05f30c2bdb2541f2836e88ab6a930cb2 : FAILURE in 9m 04s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd5e1e43d52a42f2aaa5a3f9ab02c945 : FAILURE in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f58b324fd13f09fded7005b406af5a7fcefbb546","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-14 02:30:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"20ce9b5f1a01c9c9ed3427cebe3402f0809fb8a6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-14 07:50:32.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"44bf8e8aa33d324a6b62009aa05bb183a8f6deed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 07:50:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8004f9e96bfc9424f56eff6af891fb3106175789","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 09:09:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fd01c21c3dc41c58ba73c293828785b : SUCCESS in 5m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69fb728f34d34cfbb19c4da2ffeb3ecf : SUCCESS in 5m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65994f50eb474b2ea761d380b3e56950 : SUCCESS in 5m 10s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d240542d43074c04b59ce87daa9e68df : SUCCESS in 12m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/48f7a94e06b14c70833b171226f97ec9 : SUCCESS in 14m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1885c666d0ad4e35b336d7aaf71322a2 : SUCCESS in 13m 03s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/63cce5fc3a0a4c539343cb07c6784519 : SUCCESS in 49m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/84dcf429fb1747daa0e0851cec379892 : SUCCESS in 1h 16m 35s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c070b2e401944cd3ba6c82cfca509957 : SUCCESS in 18m 51s","accounts_in_message":[],"_revision_number":2},{"id":"3b10eafb0bf011f710ed7116512a0f95816fb9c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 09:09:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0983e5888c9fbc7c8069f744abdc26b98a146fc9","revisions":{"e88e73480896c36ed50cfe2939d951173737449f":{"kind":"REWORK","_number":1,"created":"2020-07-13 19:20:20.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/740788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/88/740788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/740788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/740788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/740788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/740788/1"}}},"commit":{"parents":[{"commit":"3ac484f1d2aa0e84997cfddcb63d343cbda9a25b","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/3ac484f1d2aa0e84997cfddcb63d343cbda9a25b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-07-13 19:20:11.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-07-13 19:20:11.000000000","tz":120},"subject":"Change to relative include","message":"Change to relative include\n\nChange-Id: Iff924ab75d40ffba6a213c54a6e28933b18cf7e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/e88e73480896c36ed50cfe2939d951173737449f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/e88e73480896c36ed50cfe2939d951173737449f"}]},"branch":"refs/heads/master"},"0983e5888c9fbc7c8069f744abdc26b98a146fc9":{"kind":"REWORK","_number":2,"created":"2020-07-13 21:44:05.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/740788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/88/740788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/740788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/740788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/740788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/88/740788/2"}}},"commit":{"parents":[{"commit":"3ac484f1d2aa0e84997cfddcb63d343cbda9a25b","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/3ac484f1d2aa0e84997cfddcb63d343cbda9a25b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-07-13 19:20:11.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-07-13 21:43:57.000000000","tz":120},"subject":"Change to relative include","message":"Change to relative include\n\nChange-Id: Iff924ab75d40ffba6a213c54a6e28933b18cf7e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/0983e5888c9fbc7c8069f744abdc26b98a146fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/0983e5888c9fbc7c8069f744abdc26b98a146fc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
