)]}'
{"id":"openstack%2Fpuppet-gnocchi~747938","triplet_id":"openstack%2Fpuppet-gnocchi~stable%2Fussuri~I299640aed47071381efb4dfb95f77245223b7dce","project":"openstack/puppet-gnocchi","branch":"stable/ussuri","topic":"centos8-beaker-stable/ussuri","hashtags":[],"change_id":"I299640aed47071381efb4dfb95f77245223b7dce","subject":"Use common class to manage apache service in beaker job","status":"MERGED","created":"2020-08-25 13:31:18.000000000","updated":"2020-08-26 08:12:57.000000000","submitted":"2020-08-26 08:12:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747938-1598429577102-f4e8b138","meta_rev_id":"63d4068e107cb0cd08808c3bbeaca4789c9a8fb2","_number":747938,"virtual_id_number":747938,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-08-26 07:23:58.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 08:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-26 02:14:00.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-08-25 20:08:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-26 02:14:00.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 20:08:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-26 02:14:00.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-26 08:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c92bd3ba2a19bf6ea57e5ee8c5638851e00e77f5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-25 13:31:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5acc4cab5c744a01cf6f5f7a1eae4cf181d5e1af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 13:53:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c353b268f144c7b98e8ec71351fe84e : SUCCESS in 5m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa2233558d4446088a75d78722022956 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3dd7c77bdc9c4df49b78d32d26363013 : SUCCESS in 4m 50s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/32c7f0ef3dc54935b6eb4b7e95d4a2ff : SUCCESS in 17m 12s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/77ab2b40a08a4a08ae24760821809af3 : FAILURE in 9m 32s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9050ea21f72b4b1dae7430943a97093f : FAILURE in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0fde7919ee82c2f40e2e178240d934dd2e659d02","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-25 20:08:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"636f0181559accd11936c190e0131275958dbab6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-26 02:14:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d225d961fce5d09b318e4ab62604330b6f6381d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-26 07:23:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa4181fe7f409fb8745cdd494c1ab1d3fac2b467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 07:53:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b20037b712f543fc9db88c7b3f21e6e2 : SUCCESS in 9m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adebe5b05a5342189989d8ad34c00688 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d312c39b640542f3a0ca6d9905d112da : SUCCESS in 4m 16s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3f2f41051f104c00bc3994a52cd0962d : SUCCESS in 18m 15s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/8366a5a4306f4f7aa375037880292480 : FAILURE in 10m 02s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f7ca0cc629d4b7da3c03199d1236793 : FAILURE in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abe78bde8fc4e2b79c53ed82edc344e0aab85abd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 07:53:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4ced3d36abf40cf131162569d554640413ae374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:12:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/706e513ed3934279a833c1b2b4509744 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b41dc93940444bc19f1c3ceeb7784c5f : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2c0a9028bf541fb8487128ea83e001d : SUCCESS in 3m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9873947395ed48c09d93fbb00df3ed1f : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f7335b00517061b2ff5d33d91659bff1d00da16a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:12:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2036a5f6e543b26eb8b037f63c38a9fd110a2a8a","revisions":{"2036a5f6e543b26eb8b037f63c38a9fd110a2a8a":{"kind":"REWORK","_number":1,"created":"2020-08-25 13:31:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/747938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/38/747938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/747938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/747938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/747938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/38/747938/1"}}},"commit":{"parents":[{"commit":"911a396c08985dcac0137fba5ea99d5931d64f1c","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/911a396c08985dcac0137fba5ea99d5931d64f1c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-07 00:22:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-25 13:31:18.000000000","tz":0},"subject":"Use common class to manage apache service in beaker job","message":"Use common class to manage apache service in beaker job\n\nDepends-on: https://review.opendev.org/#/c/747935/\nChange-Id: I299640aed47071381efb4dfb95f77245223b7dce\n(cherry picked from commit f7e39c0adf05737c9084a650efb636f87ded8116)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/2036a5f6e543b26eb8b037f63c38a9fd110a2a8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/2036a5f6e543b26eb8b037f63c38a9fd110a2a8a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
