)]}'
{"id":"openstack%2Fpuppet-gnocchi~749047","triplet_id":"openstack%2Fpuppet-gnocchi~master~I47ec7e66f421045803921ce1d15c0fd69f99b2e5","project":"openstack/puppet-gnocchi","branch":"master","topic":"check-distro","hashtags":[],"change_id":"I47ec7e66f421045803921ce1d15c0fd69f99b2e5","subject":"Do not check distro names in each manifests","status":"MERGED","created":"2020-08-31 12:57:21.000000000","updated":"2020-09-01 14:00:13.000000000","submitted":"2020-09-01 14:00:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749047-1598968813068-69a35627","meta_rev_id":"4181cc9c9e0522e1696e40758d350613686eb61f","_number":749047,"virtual_id_number":749047,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 14:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-01 12:50:14.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-01 07:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-01 12:50:14.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 07:17:42.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-09-01 12:50:14.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-01 14:00:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b517525066014ddb917a784e19c234cea34c43b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 12:57:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8305d369897a828078e774bef132e7fc2b436eb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 14:08:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8e6addb1da543a580c6405f289e41c2 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d21fb49aa1849a294999eec57b03492 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/edcdfa0d5fa342eda3b859894846e24c : SUCCESS in 5m 53s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/041bf1465fae4aab8fc636c57c6428d4 : FAILURE in 12m 48s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d28d1827fb144917a718e3df1ba18dd6 : FAILURE in 13m 11s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1ac4595f1a9477e9cd779848ad2142d : FAILURE in 10m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aba52640b7b846c1b6767e7a83ed0af5 : FAILURE in 8m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/aef8b5bac29d4ce8b7504664db591bd4 : SUCCESS in 49m 15s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5b86204314864243bd7e85236a02e8e8 : SUCCESS in 1h 08m 50s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/eeaaf03393d84028a35daff129c9a836 : SUCCESS in 21m 32s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dff0c2bc7de2454c98b75d0f25ab9f61 : FAILURE in 16m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5318ee59081c33e0818f5dc0379083da217c137","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 14:21:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2354bc508a4e3188e6bc72d816afef9e2ce6a68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:55:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60d3f29c6e0047098244c11d031a1e17 : SUCCESS in 4m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9df58b99b20943a38542da6ad308de44 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bac125e2f684eaa9374011b2c8d44c7 : SUCCESS in 4m 37s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f194e5c7156c428f8c851afcb10d5193 : SUCCESS in 11m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/552d205ab77349faae4350a21bb766d5 : SUCCESS in 12m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d919e03fd98b4f4eb8a41c2e892c9976 : SUCCESS in 9m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab1374322e6d41fdb44f600fd6f9db60 : SUCCESS in 10m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1a9f04fce51046efa0241f79bb75b51c : SUCCESS in 50m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f06acdc31f7d400c8b0b1a5b5bec19e0 : SUCCESS in 1h 22m 47s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/43402275caa84c9dac35892a0552570e : SUCCESS in 19m 36s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10d90b86f6354151925bd84b7b91b368 : FAILURE in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1eb20aa76d6bca83693a7fc683ced390bb9b805a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-01 07:17:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3827920ace68c918993656bb320445cc3efffcb7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-01 12:50:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bdfde28e5f9a069fb27a3ca986ae3df4208fc8b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 12:50:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac46a262c335bc571875edfc01ed61b647cf9108","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:00:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4359f5f09bd24934bdf19a94f61ecdde : SUCCESS in 3m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a2357d84af74fa196553a5405c6c9c6 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b90e6cae446b4e67aaf7d3471c625f14 : SUCCESS in 3m 34s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/51e2a972f05949c3b08914f1312741eb : SUCCESS in 10m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e8ea2df0f51a4e299a50717a2257d90d : SUCCESS in 15m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/903ea8794c0440c68e5dd35276ca5983 : SUCCESS in 16m 50s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/35aa02f1528b44419eeb200ce326b0cd : SUCCESS in 52m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/603a6f97b1eb445699a75729e0cef856 : SUCCESS in 1h 01m 22s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/2091174c51d34a2db123072b2617eb37 : SUCCESS in 17m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1afbec384b8ccae27ff10c5dee310e3cbc6bd5fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:00:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e0ea838eb8f62e9aa14dfc9ee41a96842ba6f04","revisions":{"2c347414c44b049b0ad1462d0d063b7d407e9863":{"kind":"REWORK","_number":1,"created":"2020-08-31 12:57:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/749047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/47/749047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/47/749047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/47/749047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/47/749047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/47/749047/1"}}},"commit":{"parents":[{"commit":"f7e1857771c000c25bdc34fd3bb3174df1423fc5","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/f7e1857771c000c25bdc34fd3bb3174df1423fc5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 12:50:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 12:57:10.000000000","tz":540},"subject":"Do not check distro names in each manifests","message":"Do not check distro names in each manifests\n\nThis patch migrates the logic based on distro name to gnocchi::params,\nso that we have a single manifest maintaining distro specific behavior.\n\nChange-Id: I47ec7e66f421045803921ce1d15c0fd69f99b2e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/2c347414c44b049b0ad1462d0d063b7d407e9863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/2c347414c44b049b0ad1462d0d063b7d407e9863"}]},"branch":"refs/heads/master"},"7e0ea838eb8f62e9aa14dfc9ee41a96842ba6f04":{"kind":"REWORK","_number":2,"created":"2020-08-31 14:21:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/749047/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/47/749047/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/47/749047/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/47/749047/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/47/749047/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/47/749047/2"}}},"commit":{"parents":[{"commit":"f7e1857771c000c25bdc34fd3bb3174df1423fc5","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/f7e1857771c000c25bdc34fd3bb3174df1423fc5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 12:50:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 14:21:30.000000000","tz":540},"subject":"Do not check distro names in each manifests","message":"Do not check distro names in each manifests\n\nThis patch migrates the logic based on distro name to gnocchi::params,\nso that we have a single manifest maintaining distro specific behavior.\n\nChange-Id: I47ec7e66f421045803921ce1d15c0fd69f99b2e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7e0ea838eb8f62e9aa14dfc9ee41a96842ba6f04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/7e0ea838eb8f62e9aa14dfc9ee41a96842ba6f04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
