)]}'
{"id":"openstack%2Fpuppet-gnocchi~761071","triplet_id":"openstack%2Fpuppet-gnocchi~master~I5aa56bfced1301ba6ecc8503d7f8874f1fc67709","project":"openstack/puppet-gnocchi","branch":"master","topic":"at-params","hashtags":[],"change_id":"I5aa56bfced1301ba6ecc8503d7f8874f1fc67709","subject":"Add support for the keystone_authtoken/service_type parameter","status":"MERGED","created":"2020-11-03 08:06:20.000000000","updated":"2020-11-06 23:33:07.000000000","submitted":"2020-11-06 23:31:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761071-1604705518705-6545178d","meta_rev_id":"ada4bcb6ec728eb7c07bb64798e59583780f080f","_number":761071,"virtual_id_number":761071,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-06 23:31:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2020-11-03 09:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-11-06 22:17:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-11-06 22:17:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-03 09:43:52.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-11-06 22:17:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-06 23:31:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79849c5035d121d3e1c593d84d27d4fa28d77cc2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-03 08:06:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd3501a5feea9615858d2b7f23e0d95ec307f697","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:39:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3027350ac20943e985a0fb511b572034 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd82cb1504404cafa69ad47f3b0a3895 : SUCCESS in 3m 13s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/41959dfaa248494db41af79ef02b13db : SUCCESS in 9m 37s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1063c8c20ee547768ff9644b5c23e4c9 : SUCCESS in 7m 38s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9d487b3cbcb4035b68f438f17ca2d88 : SUCCESS in 8m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3fac22456da541d0a8a3e20c6f4344de : FAILURE in 6m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6e5f2a7398734228989d193803ab38ec : SUCCESS in 1h 21m 33s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/491e714e2ba84976916a6d54d4c996a4 : SUCCESS in 17m 35s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fd378a8406946c09dffc6f7c045fdfc : FAILURE in 11m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9233f465277d4f528f01694a4d55c91e : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b80964b00261cc7b04d8b3b121faafe370497553","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-11-03 09:43:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"291957f1c8613e8b3d880c4d8431e1db2cd617d4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-06 22:17:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e19744afeaf8ada9d6d7399fb16a3f7d4e53e022","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 22:17:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b190581e641cf773b430ce2288d97eceb6b9b396","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 23:31:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"822f3cb8dff84302f053105d3f5d71cbde51802d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 23:31:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00558f2c73bd4040a6b186e5d1402c8e : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34ba966d2be446a6becd57b052a05642 : SUCCESS in 3m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/316ca05fd7b1438d8fc4556b68ea110f : SUCCESS in 12m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a496d184b1474dcd95545c370d3977ae : SUCCESS in 9m 44s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c99a7f84336a41b08f10c36f6afac4f0 : FAILURE in 9m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/51aec3900e6b4927983974d5c0ec2d68 : SUCCESS in 1h 09m 02s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/fea73dff5683454da43a844237f1eb5e : SUCCESS in 19m 03s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd7dc6d7efbe475fbcce54837916fcf7 : FAILURE in 15m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/898d22488b4043f3b5c2c0d59368efa3 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"88b52fe49d5a4e584175e834fe4d359386298a6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 23:33:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2906a1f5bd60413a9209836fd5b3ae92 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e4e1160a635c2b1927bc74b6bd6b617ae4bcd3d","revisions":{"0e4e1160a635c2b1927bc74b6bd6b617ae4bcd3d":{"kind":"REWORK","_number":1,"created":"2020-11-03 08:06:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/761071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-gnocchi","ref":"refs/changes/71/761071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/71/761071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/71/761071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-gnocchi refs/changes/71/761071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-gnocchi refs/changes/71/761071/1"}}},"commit":{"parents":[{"commit":"426db513d6d2560a4f90891067056532c999af4a","subject":"Fix outdated files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/426db513d6d2560a4f90891067056532c999af4a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-03 08:05:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-03 08:05:58.000000000","tz":540},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: I5aa56bfced1301ba6ecc8503d7f8874f1fc67709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/0e4e1160a635c2b1927bc74b6bd6b617ae4bcd3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-gnocchi/commit/0e4e1160a635c2b1927bc74b6bd6b617ae4bcd3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
